)]}'
{"id":"openstack%2Fproject-config~932090","triplet_id":"openstack%2Fproject-config~master~Ica1970c2ac117104ea5a34e23e82ad30376588c7","project":"openstack/project-config","branch":"master","topic":"raxflex-maxservers-0","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-10-10 15:27:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica1970c2ac117104ea5a34e23e82ad30376588c7","subject":"Set raxflex max-servers to 0","status":"MERGED","created":"2024-10-10 15:07:18.000000000","updated":"2024-10-10 15:33:21.000000000","submitted":"2024-10-10 15:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932090-raxflex-maxservers-0","meta_rev_id":"d06c738f73f96c7570ef14cec1f9310890f3f417","_number":932090,"virtual_id_number":932090,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-10 15:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2024-10-10 15:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2024-10-10 15:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 15:07:45.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2024-10-10 15:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f86ba5b535ccce3a712b7736bb7b1d7ce1b54255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-10-10 15:07:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"182e4342914b533d25216f7a1bf931fca4c1d9d7","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-10-10 15:07:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39c0f166aa53fed8cca8da91f1f8f6c5f9545d9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:16:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6af989c75c34fa48146c9baa1d1ad28\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9afea7eb38a74d5ba32fcc1fb4a51563 : SUCCESS in 4m 47s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8c52f9c4c6aa4840879f367fcebf2bcc : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"481ffe9f94c0e8a74d52e57d833a6bfb3a439a3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:17:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2281ff030acab7e98d50af2f72a65a3d3ec4fb19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:27:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a588735e78b478ca868301ec8f72726\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0f876e879d024575a6f231284f3b6e1c : SUCCESS in 4m 58s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8344350164f04bbea36d9daf33a9b3ad : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5fd62fc1e83dd30b9d0be306d7d493f2dcca618a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:27:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d06c738f73f96c7570ef14cec1f9310890f3f417","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:33:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae1bc917b11d49f3b0d40bbbd2762cdf\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/852872fe381343e6ba6afbfa3ba74e40 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a91444b75702152927610d361e0a2e168a90e128","revisions":{"a91444b75702152927610d361e0a2e168a90e128":{"kind":"REWORK","_number":1,"created":"2024-10-10 15:07:18.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/90/932090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/932090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/932090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/932090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/932090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/932090/1"}}},"commit":{"parents":[{"commit":"df402816b3ffb637d9f827a56715af7e0ea83bb9","subject":"Merge \"Enable nodepool-in-zuul for opendev tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df402816b3ffb637d9f827a56715af7e0ea83bb9"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-10-10 15:06:03.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-10-10 15:06:03.000000000","tz":-420},"subject":"Set raxflex max-servers to 0","message":"Set raxflex max-servers to 0\n\nRax flex\u0027s mirror node lost its apache and openafs cache cinder volume.\nWe\u0027re getting errors like:\n\n  I/O error, dev vdd, sector 0 op 0x0:(READ) flags 0x0 phys_seg 1 prio class 0\n\nAnd it doesn\u0027t even recognize that there is an lvm partition scheme\npresent on the device. This is preventing the mirror from functioning.\n\nDrop max servers to 0 as we need a working mirror for jobs to run\nsuccessfully.\n\nChange-Id: Ica1970c2ac117104ea5a34e23e82ad30376588c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a91444b75702152927610d361e0a2e168a90e128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a91444b75702152927610d361e0a2e168a90e128"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
