)]}'
{"id":"openstack%2Fproject-config~933397","triplet_id":"openstack%2Fproject-config~master~I7d862c08b90b53c8eade6bdb5c5e674181298c73","project":"openstack/project-config","branch":"master","topic":"issue/TOBIKO-120","hashtags":[],"change_id":"I7d862c08b90b53c8eade6bdb5c5e674181298c73","subject":"DNM Just test of the new config of the trigger-readthedocs-webhook job","status":"ABANDONED","created":"2024-10-25 08:35:12.000000000","updated":"2024-11-05 10:09:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6dc93b056dfbc9b61d465fe2dcb954b550a6d0f9","_number":933397,"virtual_id_number":933397,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-25 09:28:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 08:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1baa56260a853294cf4547073c836adb4b758c11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-25 08:35:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e9686631d87857d6f35002e708db7ccb6bd0b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 08:40:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ec271edb1274ac6bc16e42396c8c2f7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/73f0f3244b1146029699941b35a5dc8e : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"18a0c5339ae6b87782caeeb7b9eedf914cbb19e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-25 09:17:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b3cb79ef418ad7c6d572818c615f3b59d4ab14f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 09:28:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d7861e6c54464aa34efc2c641be97e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d4981511fdb3422c83d157c9e5752245 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"6dc93b056dfbc9b61d465fe2dcb954b550a6d0f9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-05 10:09:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c54c6e70f306ab7f9d7e748e468be2f81c2eb385","revisions":{"5f05e0569d3633ed003979fef93ec461ef1d838c":{"kind":"REWORK","_number":1,"created":"2024-10-25 08:35:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/933397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/933397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/933397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/933397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/933397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/933397/1"}}},"commit":{"parents":[{"commit":"5a2334efdacdb876a3b9d27e0e84616e218d8b6c","subject":"Remove rtd_secret from the trigger-readthedocs-webhook job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a2334efdacdb876a3b9d27e0e84616e218d8b6c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-10-25 08:34:34.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-10-25 08:34:34.000000000","tz":120},"subject":"DNM Just test of the new config of the trigger-readthedocs-webhook job","message":"DNM Just test of the new config of the trigger-readthedocs-webhook job\n\nChange-Id: I7d862c08b90b53c8eade6bdb5c5e674181298c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f05e0569d3633ed003979fef93ec461ef1d838c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f05e0569d3633ed003979fef93ec461ef1d838c"}]},"branch":"refs/heads/master"},"c54c6e70f306ab7f9d7e748e468be2f81c2eb385":{"kind":"REWORK","_number":2,"created":"2024-10-25 09:17:00.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/933397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/933397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/933397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/933397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/933397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/933397/2"}}},"commit":{"parents":[{"commit":"64a2ca9ffa71dd090aafd2d41216e2bc679a62a2","subject":"Remove rtd_secret from the trigger-readthedocs-webhook job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64a2ca9ffa71dd090aafd2d41216e2bc679a62a2"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-10-25 08:34:34.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-10-25 09:16:43.000000000","tz":120},"subject":"DNM Just test of the new config of the trigger-readthedocs-webhook job","message":"DNM Just test of the new config of the trigger-readthedocs-webhook job\n\nChange-Id: I7d862c08b90b53c8eade6bdb5c5e674181298c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c54c6e70f306ab7f9d7e748e468be2f81c2eb385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c54c6e70f306ab7f9d7e748e468be2f81c2eb385"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
