)]}'
{"id":"openstack%2Fproject-config~935154","triplet_id":"openstack%2Fproject-config~master~Iced79ae3c13a0276ad09dae56596d825b50ae1d0","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"35994":{"account":{"_account_id":35994,"name":"Tae Park","display_name":"Tae Park","email":"Tae.Park@windriver.com","username":"tpark"},"last_update":"2024-11-14 17:10:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iced79ae3c13a0276ad09dae56596d825b50ae1d0","subject":"Add repo app-openbao for starlingx","status":"MERGED","created":"2024-11-14 14:49:48.000000000","updated":"2024-11-14 17:29:36.000000000","submitted":"2024-11-14 17:10:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935154-new-project","meta_rev_id":"ba869deeb5bf1d578c442ca8fe0a3b534b9dc12b","_number":935154,"virtual_id_number":935154,"owner":{"_account_id":35994,"name":"Tae Park","display_name":"Tae Park","email":"Tae.Park@windriver.com","username":"tpark"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 17:10:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"all":[{"value":1,"date":"2024-11-14 17:03:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-14 17:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-14 17:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 14:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 17:03:09.000000000","updated_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2024-11-14 17:06:28.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"c952e8e6917374286a7360037102e96f622275ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35994,"name":"Tae Park","display_name":"Tae Park","email":"Tae.Park@windriver.com","username":"tpark"},"date":"2024-11-14 14:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53e1304e7df75b234f090fb65648e60a90d87592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 14:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49bcf5c1fa5e43f09d98c6d7cd2a610c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/449076e03fe846078fe7203b5873e381 : SUCCESS in 4m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/faa6485316f2447ba8e58eafc65cb25d : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ec3dd18b8426216e8d732f0a12400c4ce91246a0","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2024-11-14 17:03:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"60d036a9ba599b5ce3bf7172039b570817b1e9c1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-11-14 17:06:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12f3ae658ce1c3bd22d775df680015ca7caf0b8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:06:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac97011f0daa264888055de907a51d8044551bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:10:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b34c3eb3534b59b9097c6b143eeed5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/73995ae49b714d0588019d7bcbb4115a : SUCCESS in 3m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/73fa0bf3953241dc9ca8761c81606abd : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"87db5d0ab1ad9cb4562817108c82c53f03162ea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:10:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ba869deeb5bf1d578c442ca8fe0a3b534b9dc12b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:29:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847d60c501694a60855c47398a8427cc\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/99e30594681140599b3a2de6f3770972 : SUCCESS in 3m 48s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/9e92f11202ea41fc9c940a5b393598fe : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa7196aaeba1f3801ef909346a0290e38407a687","revisions":{"aa7196aaeba1f3801ef909346a0290e38407a687":{"kind":"REWORK","_number":1,"created":"2024-11-14 14:49:48.000000000","uploader":{"_account_id":35994,"name":"Tae Park","display_name":"Tae Park","email":"Tae.Park@windriver.com","username":"tpark"},"ref":"refs/changes/54/935154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/935154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/935154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/935154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/935154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/935154/1"}}},"commit":{"parents":[{"commit":"980add8b9c1a67b36b9ec3bbfb42b1f89129fefd","subject":"Fix openstack developer docs promote job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/980add8b9c1a67b36b9ec3bbfb42b1f89129fefd"}]}],"author":{"name":"Tae Park","email":"tae.park@windriver.com","date":"2024-11-14 14:46:19.000000000","tz":-300},"committer":{"name":"Tae Park","email":"tae.park@windriver.com","date":"2024-11-14 14:49:48.000000000","tz":-300},"subject":"Add repo app-openbao for starlingx","message":"Add repo app-openbao for starlingx\n\nThis commit adds a new repo named app-openbao, an openbao application\nrepo for StarlingX\n\nChange-Id: Iced79ae3c13a0276ad09dae56596d825b50ae1d0\nSigned-off-by: Tae Park \u003ctae.park@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa7196aaeba1f3801ef909346a0290e38407a687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa7196aaeba1f3801ef909346a0290e38407a687"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
