)]}'
{"id":"openstack%2Fproject-config~936754","triplet_id":"openstack%2Fproject-config~master~I3f5e4b9d148ec01a6603458a8e9abba2c2e57152","project":"openstack/project-config","branch":"master","topic":"watcher-stable-branch-config","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-12-19 01:45:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f5e4b9d148ec01a6603458a8e9abba2c2e57152","subject":"update watcher repos to use -release group for stable branch","status":"MERGED","created":"2024-11-29 14:48:59.000000000","updated":"2024-12-19 01:50:48.000000000","submitted":"2024-12-19 01:45:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936754-watcher-stable-branch-config","meta_rev_id":"0ecbf9547611f6ace93aad455d5ac308ff695fb2","_number":936754,"virtual_id_number":936754,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 01:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-12-19 01:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-11-29 15:09:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-01 23:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-12-19 01:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-29 15:00:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-29 15:09:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-12-01 23:48:39.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2024-12-19 01:31:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"9761e954a35890f4db5f87d94c94b3b831758c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-29 14:48:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e05e6910a1b4d5ed3e0ff8f8d7cb5b064b814938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 15:00:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4294465c8b0438991ee5da11b42a5f0\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/536378bb02eb4a659c8a7a87a1ca1066 : SUCCESS in 4m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/38c1aea5c30a4f908c3ad0bf2a9989ed : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"086b7d4970570202d64a115ad2b7db1918755f36","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-29 15:09:29.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6f25f8529aece57f186755b7e5f57f47526ae41","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2024-12-01 23:48:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e2fa46e180bf7e2d4e0b6ab9457022d8c488c71","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-04 18:40:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"527e1344401e1d23642871d6cb19047ef049339e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-12-19 01:31:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5050b3b422ae05c5299699426f93f0f9a1ca847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:31:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d5945df311c0bd479c9a4b594423f09660e876b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:45:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5da8c68aee49a5a6be50f6498c80b2\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0f76919b2e5b467aa43d6dd1a7ff583e : SUCCESS in 2m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bfd58845e312468b89d4d485cbcff99c : SUCCESS in 4m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2ea65583d37641c59fbdc1bff77e29cf : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"abf99e7fa5e163ceeb95c683e74cb9d1885c50c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:45:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ecbf9547611f6ace93aad455d5ac308ff695fb2","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:50:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f135edc2d1a481fa438b4f40590d0db\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f32d463d7b084f9685c720794d7e1712 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f5374c73fbb4dff2017a03663d8cd3a5f5748cc","revisions":{"2f5374c73fbb4dff2017a03663d8cd3a5f5748cc":{"kind":"REWORK","_number":1,"created":"2024-11-29 14:48:59.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/54/936754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/936754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/936754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/936754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/936754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/936754/1"}}},"commit":{"parents":[{"commit":"04d6e0c87787f5586edfadc4f46e09fabc75c5d5","subject":"Disable docker hub mirror use in jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04d6e0c87787f5586edfadc4f46e09fabc75c5d5"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-29 14:43:18.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-29 14:48:54.000000000","tz":0},"subject":"update watcher repos to use -release group for stable branch","message":"update watcher repos to use -release group for stable branch\n\nWatcher has had watcher-release and watcher-dashboard-release\nfor a long time but they have not been correctly configured\nin the gerrit ACLs.\n\nThis change updates the relevent acl config to allow\nwatcher-release member to merge patches on stable branches for\nwatcher and python-watcherclint, and watcher-dashboard-release\nto manage the watcher-dashboard stable branches.\n\nThe stable-maint-core group is a member of both -release groups\nand will inherit review rights as a result of this change.\n\nThe watcher tempest and spec configs are not updated as they\nare not branched.\n\nChange-Id: I3f5e4b9d148ec01a6603458a8e9abba2c2e57152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f5374c73fbb4dff2017a03663d8cd3a5f5748cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f5374c73fbb4dff2017a03663d8cd3a5f5748cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
