)]}'
{"id":"openstack%2Fproject-config~937339","triplet_id":"openstack%2Fproject-config~master~I9d392cddc73d16531f151d13cdbb6b81cf70a042","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2024-12-09 13:29:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d392cddc73d16531f151d13cdbb6b81cf70a042","subject":"Add openstack-k8s-operators/sg-core to zuul repos","status":"MERGED","created":"2024-12-09 09:52:22.000000000","updated":"2024-12-09 13:37:02.000000000","submitted":"2024-12-09 13:29:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937339","meta_rev_id":"8f836a8b2bcf84a4f3b8da4c46041d7e7ee266bc","_number":937339,"virtual_id_number":937339,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-09 13:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-09 13:06:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-09 13:06:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-09 10:02:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 13:06:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"3ac9b619611aadabd8273dbf1d73d772c7666417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-12-09 09:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4e24040fb1834ce589198c3309b46c3d34de58b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 10:02:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5822e030dd28439f860e93ff28300fa7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ddc310ead0014ecaa1fb054c65dc6154 : SUCCESS in 4m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/08157a6f8d844e85beebdacaedbf8522 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6b72e6a790b9fc2424d1f5ee3a9c8b411497b5e1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-12-09 13:06:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c9a8e084e2558f0325b7e78c31dad4a5b6a8ba0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:07:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43bfcd14a3b1efc75d70ac6de882d1d62d992fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:29:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5615b5012e643b799bc310f4e99d809\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/35799a14edc042c8a7376e9852f403d9 : SUCCESS in 4m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9fac95f739a24354a2b4974fd993cf4b : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"96f0b7ca1bde701a3d41b88ecc3eaa668d550d06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:29:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f836a8b2bcf84a4f3b8da4c46041d7e7ee266bc","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:37:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19760096b0304b0da7d2fee734a8c816\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/771a63aba99945dd89bc811361c33ed6 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f","revisions":{"b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f":{"kind":"REWORK","_number":1,"created":"2024-12-09 09:52:22.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/39/937339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/39/937339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/39/937339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/39/937339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/39/937339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/39/937339/1"}}},"commit":{"parents":[{"commit":"d64444ef5c38d0631fb8df753a47cec5a7daaed7","subject":"Update jeepyb triggered Gerrit builds to Gerrit 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d64444ef5c38d0631fb8df753a47cec5a7daaed7"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-12-09 09:43:25.000000000","tz":60},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-12-09 09:52:20.000000000","tz":60},"subject":"Add openstack-k8s-operators/sg-core to zuul repos","message":"Add openstack-k8s-operators/sg-core to zuul repos\n\nsg-core was moved from github.com/infrawatch/sg-core to\ngithub.com/openstack-k8s-operators/sg-core. While the infrawatch\nsg-core will stay maintained for the forseeable future, the\nopenstack-k8s-operators version of sg-core will receive all\nthe new code. We\u0027d want to switch to testing with\nthe openstack-k8s-operators sg-core.\n\nOnce this merges, I\u0027ll follow-up with a patch to use\nopenstack-k8s-operators/sg-core in the telemetry-tempest-plugin\nand finally I\u0027ll propose a patch to remove the old infrawatch/sg-core\nfrom project-config.\n\nChange-Id: I9d392cddc73d16531f151d13cdbb6b81cf70a042\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
