)]}'
{"id":"openstack%2Fproject-config~937488","triplet_id":"openstack%2Fproject-config~master~Ieaf3a7ba281c3af934fa91a96eecd4654208c701","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-12-10 23:44:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieaf3a7ba281c3af934fa91a96eecd4654208c701","subject":"Pause openEuler-22-03-LTS image builds","status":"MERGED","created":"2024-12-10 22:40:26.000000000","updated":"2024-12-10 23:49:56.000000000","submitted":"2024-12-10 23:44:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"937488","meta_rev_id":"4d3aaa277b291bfc7b16e4234f9a7bc5f5f73f43","_number":937488,"virtual_id_number":937488,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 23:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-12-10 22:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2024-12-10 22:47:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36495,"name":"tzing","display_name":"tzing","email":"tzing0013@gmail.com","username":"tzing"}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-10 22:47:47.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-12-10 22:47:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":36495,"name":"tzing","display_name":"tzing","email":"tzing0013@gmail.com","username":"tzing"},"state":"CC"},{"updated":"2024-12-10 23:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ae83b581674dac4b4f1b1c1d45449fa2a1381f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-12-10 22:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5974428c8897b99aa3fcf3a4d26793a0fb03300e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-12-10 22:44:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31756c53b72e4f72ac5cf8da9deb39212d25f151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-12-10 22:46:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e21dfc6d461d42c9f5b874404cba4dcd8c488b3a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-10 22:47:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05c8878e51b4b3f61571a350afa09ba86d95c37f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-12-10 22:47:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee1e36093c4f337607c717d077dba93441d1cddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:24:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe291f3c6124d529bf76ce38e8c33ee\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a25031f6b19d486eab28d46f74bf60db : SUCCESS in 2m 56s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4856c78f445f4f8b9232eef2f24665f6 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"36d91088ed5388dfcaa34f0edfbc5efeeaa29e27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:24:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"efc86339fe9a87def82a221e02f7580039e74a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:44:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe36a505d6648e993b3f8fc20846a70\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0c8266d037cd4d67a1ba497c137d5df9 : SUCCESS in 4m 37s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/159fd2d412fb432e9e28bd1b1e22de03 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"215b98b793d803b4691fd3be152512e69538a098","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:44:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4d3aaa277b291bfc7b16e4234f9a7bc5f5f73f43","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:49:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac8fc9734dc492eb3851b90fb24533a\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/b8892622a8074891a6278cee6c4643fb : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ecf403c15b31fab69622e76bbef51bfcc458d6b7","revisions":{"64a275c8a13af61b98f82cfbf1a1bccbaeee4ff4":{"kind":"REWORK","_number":1,"created":"2024-12-10 22:40:26.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/88/937488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/937488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/937488/1"}}},"commit":{"parents":[{"commit":"b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f","subject":"Add openstack-k8s-operators/sg-core to zuul repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-12-10 22:37:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-12-10 22:37:02.000000000","tz":0},"subject":"Pause openEuler-22-03-LTS image builds","message":"Pause openEuler-22-03-LTS image builds\n\nEver since Ifcee064959972bf871688baeff7b3288d7517da1 updated the\nopenEuler mirror in OpenDev to a newer version, image builds have\nbeen failing. Pause them for now until they can be made working\nagain (presumably by switching to build a newer version of\nopenEuler).\n\nRelated, functional testing for the openeuler-minimal/22.03-LTS\nelement in diskimage-builder is being temporarily suspended by\nI40564fb93aed339bc085daf1bd157a5b11218d9a for similar reasons.\n\nChange-Id: Ieaf3a7ba281c3af934fa91a96eecd4654208c701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64a275c8a13af61b98f82cfbf1a1bccbaeee4ff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64a275c8a13af61b98f82cfbf1a1bccbaeee4ff4"}]},"branch":"refs/heads/master"},"82d4ef848dbfdcbdb82f191da35c22b5745fd86c":{"kind":"REWORK","_number":2,"created":"2024-12-10 22:44:59.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/88/937488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/937488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/937488/2"}}},"commit":{"parents":[{"commit":"b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f","subject":"Add openstack-k8s-operators/sg-core to zuul repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-12-10 22:37:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-12-10 22:44:46.000000000","tz":0},"subject":"Pause openEuler-22-03-LTS image builds","message":"Pause openEuler-22-03-LTS image builds\n\nEver since Ifcee064959972bf871688baeff7b3288d7517da1 updated the\nopenEuler mirror in OpenDev to a newer version, image builds have\nbeen failing. Pause them for now until they can be made working\nagain (presumably by switching to build a newer version of\nopenEuler).\n\nRelated, functional testing for the openeuler-minimal/22.03-LTS\nelement in diskimage-builder is being temporarily suspended by\nI40564fb93aed339bc085daf1bd157a5b11218d9a for similar reasons.\n\nChange-Id: Ieaf3a7ba281c3af934fa91a96eecd4654208c701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82d4ef848dbfdcbdb82f191da35c22b5745fd86c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82d4ef848dbfdcbdb82f191da35c22b5745fd86c"}]},"branch":"refs/heads/master"},"ecf403c15b31fab69622e76bbef51bfcc458d6b7":{"kind":"REWORK","_number":3,"created":"2024-12-10 22:46:03.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/88/937488/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/937488/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/937488/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/937488/3"}}},"commit":{"parents":[{"commit":"b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f","subject":"Add openstack-k8s-operators/sg-core to zuul repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7a7d2b06c97471fa8f49ed274b0bb270cc96c1f"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-12-10 22:37:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-12-10 22:45:57.000000000","tz":0},"subject":"Pause openEuler-22-03-LTS image builds","message":"Pause openEuler-22-03-LTS image builds\n\nEver since Ifcee064959972bf871688baeff7b3288d7517da1 updated the\nopenEuler mirror in OpenDev to a newer version, image builds have\nbeen failing. Pause them for now until they can be made working\nagain (presumably by switching to build a newer version of\nopenEuler).\n\nRelated, functional testing for the openeuler-minimal/22.03-LTS\nelement in diskimage-builder is being temporarily suspended by\nI40564fb93aed339bc085daf1bd157a5b11218d9a for similar reasons.\n\nChange-Id: Ieaf3a7ba281c3af934fa91a96eecd4654208c701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecf403c15b31fab69622e76bbef51bfcc458d6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecf403c15b31fab69622e76bbef51bfcc458d6b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
