)]}'
{"id":"openstack%2Fproject-config~937792","triplet_id":"openstack%2Fproject-config~master~Ib31902562d61516f57ebbf9247a902e9dcc80fec","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-12-17 17:13:21.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-17 09:17:09.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-12-17 17:20:30.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-12-17 17:20:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib31902562d61516f57ebbf9247a902e9dcc80fec","subject":"Authorize packstack-release to delete","status":"MERGED","created":"2024-12-16 16:06:49.000000000","updated":"2024-12-17 17:26:22.000000000","submitted":"2024-12-17 17:20:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937792","meta_rev_id":"ab5196144735f36339dcb5be6cece1ba572178d3","_number":937792,"virtual_id_number":937792,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 17:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-12-17 17:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2024-12-17 09:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2024-12-17 17:13:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-16 16:09:40.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-12-16 16:09:40.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-12-16 16:09:40.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-12-16 16:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"178a0def145dba606b884b4fa43c39da69dcbd79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 16:06:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2736e9cfb23aebdba9e7d630ea8fbbe361c22ff6","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 16:09:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7edd0adaa99d5b23d275afd36f22720d7b15b4ed","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-16 16:10:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"150ac334e9b83bc1a6e53764afbd01c1b4a8e7e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:13:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa6114f06ff54810af4ee3a89cc65ff4\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/207054b73a9b4c7ab3fe6fc40165d2cb : SUCCESS in 4m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dd19358ab80b4419914323b47c3419d6 : SUCCESS in 4m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/73403dd5f6d44759b3c0270f00890280 : FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5e1119e3cc01c953f93cfd8183ea2d158b43bf8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 16:17:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c60cd5350854f684f1948990a68af50f1d882f80","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 16:17:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2eb93955d9f38670f1ac5d56258bd086793a86d2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-16 16:22:13.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f632247e25ed9059e95fa2d205f90efc150e6621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:25:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/267c93617bb5485990751d1e8b390ce3\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a2473e062f324518bd86d12c826643ec : SUCCESS in 4m 33s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c4d5baa13ec7497f938f8905b34c4ebf : SUCCESS in 5m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dd2f540c064d47639ba30e787dd11239 : FAILURE in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d408aa7af8ccd040f4a43a04c0470cbe2ad4879a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 17:11:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e75e47a7c59c8bf70be81d9f543af0c4fdc1e026","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 17:15:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7940ce463255879ae263f66c49fa6615f3c98c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 17:16:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e842912e2014edbb99e4c232404b01f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/113791a998ce414c92548d8b79f594fb : SUCCESS in 2m 40s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d4f9f1d342824a1aadd5a9ec9dec1189 : SUCCESS in 4m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/53b634c6f4b64b32a7c79d022ac00f31 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"1a5c1730140e58da5c1b789bd7ea8746218cdd9b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-12-16 17:57:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"74753c1adff7ac4dc9a14e27ad06ef8da8660dd1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-17 06:50:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a8222631df7cc6256cad186c4e9d6e5d24be6ea5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-17 08:27:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f2db5aa8cdb585fa240e821738d4e36ff82a085c","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-17 08:28:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"451606cf267d217a20c68220a4d90637d043b9a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 08:37:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144ecd413cee4197b32b30332d664f6c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1cd9d1a50a90480fbb616c941de7f569 : SUCCESS in 4m 35s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/767914ae0cba41e88749e8cf8cc20fa4 : SUCCESS in 4m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/52019a3e1f74427b809f452b76f8a213 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":4},{"id":"96f57c5dbe6b946cd68b2276e18d04ddcdf07ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-17 08:41:43.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3cd0b3082ce0dad874dc79c990b6f9455aba81be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 08:49:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ffaaf2b3e0c4d0389f14be5c8ab1f95\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ae5ce24606344bb69ef2a1c792164d5d : SUCCESS in 4m 05s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cdee46d7574b4a81a2a74a4fe5ac9a09 : SUCCESS in 5m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d3c226f485f4087ae4a2eaec679c7da : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":5},{"id":"4481a1f1e0eb09ab43f3be02d580a7fac98e5f2c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-17 09:17:09.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ac80bad250d57f88e2e59c56150ea642662c34da","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-17 17:13:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"908cea9b092f6dbe8f715d9d0c2f9828ebf015cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:13:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5a8cfb10bef95e48dff4098a7e76b61b991ece8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:20:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55526815f04b4e3dbf70d265b1ca8b5e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fa604e4c2575464d940d53f2e748590b : SUCCESS in 2m 40s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/404947531c714eabadf227a190c44219 : SUCCESS in 5m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f11f8c984c074bd7ac820f66fe3ddc45 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"ced38e4d0858eaf56c63ae3dd7cb4f95606b84e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:20:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ab5196144735f36339dcb5be6cece1ba572178d3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:26:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7015c0ce29b41189fc4c289d6287dfb\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f5fa40872e1e46f4bf8b7078163e0536 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6553f6c50ed34e6fb57e699a3920f4f67e9a9467","revisions":{"681b75358bc5ce6d30944cb8c7dad0454109271a":{"kind":"REWORK","_number":1,"created":"2024-12-16 16:06:49.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/92/937792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/937792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/937792/1"}}},"commit":{"parents":[{"commit":"ecf403c15b31fab69622e76bbef51bfcc458d6b7","subject":"Pause openEuler-22-03-LTS image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecf403c15b31fab69622e76bbef51bfcc458d6b7"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 16:06:10.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 16:06:12.000000000","tz":60},"subject":"Authorize packstack-core to force push to remove branch","message":"Authorize packstack-core to force push to remove branch\n\nWe need this authorization to be able to remove old branches.\n\nChange-Id: Ib31902562d61516f57ebbf9247a902e9dcc80fec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/681b75358bc5ce6d30944cb8c7dad0454109271a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/681b75358bc5ce6d30944cb8c7dad0454109271a"}]},"branch":"refs/heads/master"},"f4d084cbc08310333c0f84998761553d43c5a989":{"kind":"REWORK","_number":2,"created":"2024-12-16 16:17:33.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/92/937792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/937792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/937792/2"}}},"commit":{"parents":[{"commit":"ecf403c15b31fab69622e76bbef51bfcc458d6b7","subject":"Pause openEuler-22-03-LTS image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecf403c15b31fab69622e76bbef51bfcc458d6b7"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 16:06:10.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 16:17:27.000000000","tz":60},"subject":"Authorize packstack-core to force push to remove branch","message":"Authorize packstack-core to force push to remove branch\n\nWe need this authorization to be able to remove old branches.\n\nChange-Id: Ib31902562d61516f57ebbf9247a902e9dcc80fec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f4d084cbc08310333c0f84998761553d43c5a989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f4d084cbc08310333c0f84998761553d43c5a989"}]},"branch":"refs/heads/master"},"cece80ba49b77b5054fd9fcf6eddb6c692c438bb":{"kind":"REWORK","_number":3,"created":"2024-12-16 17:11:54.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/92/937792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/937792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/937792/3"}}},"commit":{"parents":[{"commit":"ecf403c15b31fab69622e76bbef51bfcc458d6b7","subject":"Pause openEuler-22-03-LTS image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecf403c15b31fab69622e76bbef51bfcc458d6b7"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 16:06:10.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 17:11:50.000000000","tz":60},"subject":"Authorize packstack-core to force push to remove branch","message":"Authorize packstack-core to force push to remove branch\n\nWe need this authorization to be able to remove old branches.\n\nChange-Id: Ib31902562d61516f57ebbf9247a902e9dcc80fec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cece80ba49b77b5054fd9fcf6eddb6c692c438bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cece80ba49b77b5054fd9fcf6eddb6c692c438bb"}]},"branch":"refs/heads/master"},"5fec13d5b07e5d8546a7712cb3e65fad3af25528":{"kind":"REWORK","_number":4,"created":"2024-12-17 08:27:37.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/92/937792/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/937792/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/937792/4"}}},"commit":{"parents":[{"commit":"ecf403c15b31fab69622e76bbef51bfcc458d6b7","subject":"Pause openEuler-22-03-LTS image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecf403c15b31fab69622e76bbef51bfcc458d6b7"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 16:06:10.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-17 08:27:32.000000000","tz":60},"subject":"Authorize packstack-core to force push to remove branch","message":"Authorize packstack-core to force push to remove branch\n\nWe need this authorization to be able to remove old branches.\n\nChange-Id: Ib31902562d61516f57ebbf9247a902e9dcc80fec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5fec13d5b07e5d8546a7712cb3e65fad3af25528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5fec13d5b07e5d8546a7712cb3e65fad3af25528"}]},"branch":"refs/heads/master"},"6553f6c50ed34e6fb57e699a3920f4f67e9a9467":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-12-17 08:41:43.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/92/937792/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/937792/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/937792/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/937792/5"}}},"commit":{"parents":[{"commit":"ecf403c15b31fab69622e76bbef51bfcc458d6b7","subject":"Pause openEuler-22-03-LTS image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecf403c15b31fab69622e76bbef51bfcc458d6b7"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 16:06:10.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-17 08:41:20.000000000","tz":60},"subject":"Authorize packstack-release to delete","message":"Authorize packstack-release to delete\n\nWe need this authorization to be able to remove old branches.\n\nChange-Id: Ib31902562d61516f57ebbf9247a902e9dcc80fec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6553f6c50ed34e6fb57e699a3920f4f67e9a9467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6553f6c50ed34e6fb57e699a3920f4f67e9a9467"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
