)]}'
{"id":"openstack%2Fproject-config~938537","triplet_id":"openstack%2Fproject-config~master~I9bf95b062fb6cf290ef8d86340ab9823907200c9","project":"openstack/project-config","branch":"master","topic":"use-oko-sgcore","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-01-16 10:10:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9bf95b062fb6cf290ef8d86340ab9823907200c9","subject":"Remove infrawatch/sg-core from zuul repos","status":"MERGED","created":"2025-01-07 08:11:14.000000000","updated":"2025-01-16 10:30:27.000000000","submitted":"2025-01-16 10:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938537-use-oko-sgcore","meta_rev_id":"84955421cf9e53bc9d188cf58f09ce3f07fea0c5","_number":938537,"virtual_id_number":938537,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 10:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-01-07 13:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-16 09:59:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 08:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 13:52:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"01e0634ee4c6b795a2653d62ccd05b346498c46a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-07 08:11:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f7bb0568c09d0972f183b2c39f9a865aeab3379","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-07 08:11:46.000000000","message":"Topic set to use-oko-sgcore","accounts_in_message":[],"_revision_number":1},{"id":"e7df9f7054fadc7db17577966c4616dea380705b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 08:34:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87ec891d88944dea9337fcbf0e9db44e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1ff71df4ae9145c6ac15921a8a2b65f5 : SUCCESS in 3m 09s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b1307fcd7fd242d58147c033ffa1e208 : SUCCESS in 4m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/76a091e7f6ec4d6d8f4a24283ef6ae5d : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"062a3436c108361ecf145c32c3e6e51c6e7d47e5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-07 13:52:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2bbb842040ffdbebfa8d9863a37e34bb774dd622","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-16 09:59:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b27ea7eba9efcdb4b6044143c6c3d37466732cc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 10:00:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69645ba768bd51d6d1a4238a85b9a320d2e4e355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 10:10:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac13aa33470b4a15bb6591afc6b97fb3\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6bfff800462d4298baad9a83a8f9f995 : SUCCESS in 3m 29s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d1008b010524ba0a1e785b5eedae070 : SUCCESS in 5m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0104a77787e1422ba3268e94d0a1695d : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c64f5abe6c1665d4f3c9df84b908e70814abb003","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 10:10:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84955421cf9e53bc9d188cf58f09ce3f07fea0c5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 10:30:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95285ba2afcb46108a7f0896149dacd9\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/ad7148403f8b45a99d7236c690706c95 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e33bd93d7f947c0c84e221391f0a1553908f9a98","revisions":{"e33bd93d7f947c0c84e221391f0a1553908f9a98":{"kind":"REWORK","_number":1,"created":"2025-01-07 08:11:14.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/37/938537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/938537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/938537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/938537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/938537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/938537/1"}}},"commit":{"parents":[{"commit":"c7b396f48c6c5da627d63bc1d39918358534fd7f","subject":"Run periodic wheel cache publishing only on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7b396f48c6c5da627d63bc1d39918358534fd7f"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-01-07 08:08:32.000000000","tz":60},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-01-07 08:08:32.000000000","tz":60},"subject":"Remove infrawatch/sg-core from zuul repos","message":"Remove infrawatch/sg-core from zuul repos\n\ninfrawatch/sg-core isn\u0027t being used anywhere anymore, so it can\nbe removed from the project-config. It was recently replaced\nby openstack-k8s-operators/sg-core.\n\nChange-Id: I9bf95b062fb6cf290ef8d86340ab9823907200c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e33bd93d7f947c0c84e221391f0a1553908f9a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e33bd93d7f947c0c84e221391f0a1553908f9a98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
