)]}'
{"id":"openstack%2Fproject-config~938908","triplet_id":"openstack%2Fproject-config~master~I4971690bcf4db7ac56dda271153560458c07f753","project":"openstack/project-config","branch":"master","topic":"rv_priority_for_neutron_projects","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-27 13:41:31.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-02-27 13:41:31.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-01-16 16:37:59.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I4971690bcf4db7ac56dda271153560458c07f753","subject":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp","status":"MERGED","created":"2025-01-10 14:45:51.000000000","updated":"2025-02-27 13:45:47.000000000","submitted":"2025-02-27 13:41:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938908-rv_priority_for_neutron_projects","meta_rev_id":"2f006d129764d7a5ccd90f45d1f558d47540f654","_number":938908,"virtual_id_number":938908,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-27 13:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2025-02-27 09:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-02-27 09:17:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-02-27 09:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-27 13:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-27 13:28:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 14:51:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-01-10 14:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 14:54:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-01-10 16:18:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-01-16 09:56:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-01-16 09:56:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-02-27 13:28:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"fc5e4af89379583bbcc1fe74fdf32c2e83ab18ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 14:45:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91a8cb21812f5238db5cc43a22634e56347009f0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-10 14:51:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7bb5c0cebf809771db6b4f3435a962ebae2716c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 14:54:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac39b7d002f348788d215c9478b6e0c9\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e1aec4b3d9b045f0854a8c6fbbc378af : SUCCESS in 3m 29s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c9a0938bea7a4b97b4595256852fbaa5 : SUCCESS in 4m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d77ac2f3f0f3476b80b33fc90423c5a8 : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0789985d96ca39af0f8ccd6e44af064f809f33f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 14:54:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"21c41ccc925c96288355e13fbc4e2a8d131df7c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 14:55:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"195d677899eb76d0595b62f9901dadacade315a3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-10 15:04:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d5c88ea9eadc7a3dd9e4383fe237e2fc20ab228c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 15:05:38.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"437e375c0c37346ee1a6bca6ae0ba15e30325fa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-10 15:09:43.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"582c9edfd1f7b9b2c94bd2fa46d3545a7b18307f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:35:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/788a293e9b33431aacec49ecd4a0ef7a\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/20e670c4f6404210973aa370d119d479 : SUCCESS in 3m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a6e35f296cf44d5aa263685c3181acb9 : SUCCESS in 4m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d17a45a892b4451ae07ccb6692a0de2 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":5},{"id":"e409bae10b1640650e3e8e27ed57ca942a0db262","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-10 16:18:44.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c61ee5809ed3d8e0e7a0e287de6b39ce1be35e39","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-16 09:56:48.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c069969296ec1849a1fc0cc51f8d24735d8ebec5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-16 16:37:59.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d0f5cb7702286c05561606e51472a69a3f8a7b5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-27 09:17:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f30f503223a7c5e1ecb56a345d1d79222e8bd00f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 10:23:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2f235593464b3ca370dd30f3452fc9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f44b1c6520434f44bf4be36d391ade94 : SUCCESS in 4m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b173a120400d47d5aad8491e3d8f0190 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":6},{"id":"949c8227965665d92e05d4f42bde86eb20fcc594","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-27 13:28:56.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7a30e7585ddfea4d84caf69eef196b904e41d847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 13:29:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e3a6fe795fbcc140f52f8f8a8b078a3d1bd0cc5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 13:41:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d2d1016d23b42d6a116d16ae43f7758\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/38d9160cd15d4218b302f444165d4107 : SUCCESS in 4m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6f0d64b08e5d4d1288264738aa61af2e : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":6},{"id":"9120e07d03797747becc2c1f711fc81b82db6fe5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 13:41:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"2f006d129764d7a5ccd90f45d1f558d47540f654","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 13:45:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d437599bbd60447a8eb37b395e83341f\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/4192a1662b0e4b80ba3b9d27c96a27a6 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e8bea5b5feaed230b193faeb27d5f6318726c5fc","revisions":{"e2ba3c280fe3811d2a51261179dbba578fe42bbe":{"kind":"REWORK","_number":1,"created":"2025-01-10 14:45:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/938908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/938908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/938908/1"}}},"commit":{"parents":[{"commit":"64386d91b625a7d65866e78cdae8f76df146e6fb","subject":"Merge \"Remove Murano/Senlin/Sahara OSA roles from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64386d91b625a7d65866e78cdae8f76df146e6fb"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:45:12.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:45:12.000000000","tz":60},"subject":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp","message":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp\n\nChange-Id: I4971690bcf4db7ac56dda271153560458c07f753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2ba3c280fe3811d2a51261179dbba578fe42bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2ba3c280fe3811d2a51261179dbba578fe42bbe"}]},"branch":"refs/heads/master"},"29d7695ca769ef6af5c771364d68b9744fa29e40":{"kind":"REWORK","_number":2,"created":"2025-01-10 14:54:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/938908/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/938908/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/938908/2"}}},"commit":{"parents":[{"commit":"64386d91b625a7d65866e78cdae8f76df146e6fb","subject":"Merge \"Remove Murano/Senlin/Sahara OSA roles from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64386d91b625a7d65866e78cdae8f76df146e6fb"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:45:12.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:52:42.000000000","tz":60},"subject":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp","message":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp\n\nChange-Id: I4971690bcf4db7ac56dda271153560458c07f753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/29d7695ca769ef6af5c771364d68b9744fa29e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/29d7695ca769ef6af5c771364d68b9744fa29e40"}]},"branch":"refs/heads/master"},"22035c961f0b5f81131f4ee8d5c0822220141d2b":{"kind":"REWORK","_number":3,"created":"2025-01-10 14:55:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/938908/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/938908/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/938908/3"}}},"commit":{"parents":[{"commit":"64386d91b625a7d65866e78cdae8f76df146e6fb","subject":"Merge \"Remove Murano/Senlin/Sahara OSA roles from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64386d91b625a7d65866e78cdae8f76df146e6fb"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:45:12.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-01-10 14:55:44.000000000","tz":0},"subject":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp","message":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp\n\nChange-Id: I4971690bcf4db7ac56dda271153560458c07f753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22035c961f0b5f81131f4ee8d5c0822220141d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22035c961f0b5f81131f4ee8d5c0822220141d2b"}]},"branch":"refs/heads/master"},"2e76507584927e5cd17c30e4949d904b6b5476ae":{"kind":"REWORK","_number":4,"created":"2025-01-10 15:05:38.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/938908/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/938908/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/938908/4"}}},"commit":{"parents":[{"commit":"64386d91b625a7d65866e78cdae8f76df146e6fb","subject":"Merge \"Remove Murano/Senlin/Sahara OSA roles from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64386d91b625a7d65866e78cdae8f76df146e6fb"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:45:12.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-01-10 15:05:38.000000000","tz":0},"subject":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp","message":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp\n\nChange-Id: I4971690bcf4db7ac56dda271153560458c07f753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2e76507584927e5cd17c30e4949d904b6b5476ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2e76507584927e5cd17c30e4949d904b6b5476ae"}]},"branch":"refs/heads/master"},"05e60643b3e68d3a983cfb8160f69f74ae5c9be2":{"kind":"REWORK","_number":5,"created":"2025-01-10 15:09:43.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/938908/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/938908/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/938908/5"}}},"commit":{"parents":[{"commit":"64386d91b625a7d65866e78cdae8f76df146e6fb","subject":"Merge \"Remove Murano/Senlin/Sahara OSA roles from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64386d91b625a7d65866e78cdae8f76df146e6fb"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:45:12.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-01-10 15:09:42.000000000","tz":0},"subject":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp","message":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp\n\nChange-Id: I4971690bcf4db7ac56dda271153560458c07f753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05e60643b3e68d3a983cfb8160f69f74ae5c9be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05e60643b3e68d3a983cfb8160f69f74ae5c9be2"}]},"branch":"refs/heads/master"},"e8bea5b5feaed230b193faeb27d5f6318726c5fc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-02-27 09:17:55.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/938908/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/938908/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/938908/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/938908/6"}}},"commit":{"parents":[{"commit":"1147a32da46c293195d478fa58b17f1c02590f7f","subject":"Update StarlingX docs promote job for R10 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1147a32da46c293195d478fa58b17f1c02590f7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-10 14:45:12.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-02-27 09:17:55.000000000","tz":0},"subject":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp","message":"Neutron: add review priority to ovn-bgp-agent and ovsdbapp\n\nChange-Id: I4971690bcf4db7ac56dda271153560458c07f753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8bea5b5feaed230b193faeb27d5f6318726c5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8bea5b5feaed230b193faeb27d5f6318726c5fc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
