)]}'
{"id":"openstack%2Fproject-config~939282","triplet_id":"openstack%2Fproject-config~master~If073f91089d19ae0ad8815f6b6e5049e711bbd0a","project":"openstack/project-config","branch":"master","topic":"ironic-lib-deprecation","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-02-24 19:16:18.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-02-24 19:16:18.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-17 01:59:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"If073f91089d19ae0ad8815f6b6e5049e711bbd0a","subject":"Deprecate ironic-lib","status":"MERGED","created":"2025-01-14 21:25:33.000000000","updated":"2025-02-24 19:16:18.000000000","submitted":"2025-02-24 19:16:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"939282-ironic-lib-deprecation","meta_rev_id":"cba843ebe1a5a6bada6fe72694a70bfa93f3fdfa","_number":939282,"virtual_id_number":939282,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-24 19:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-02-17 10:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-17 01:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2025-01-16 22:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2025-02-24 19:09:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 21:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 21:58:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-01-14 22:04:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-01-15 18:19:15.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-02-17 10:09:55.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"1317ea79ab0a8ddc1ffe5ad50533d8017439691a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:25:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73bc9c1482fc8cd4b2e6cea3dbdc2320dc51fd31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 21:26:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d32fb55bc1b42aa8017167b593920d7\n\n\nWarning:\n  Dependency cycle detected and project openstack/ironic-lib doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":1},{"id":"0b39aa6b295e30c81b8b5289e09f1424e899eb28","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:26:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07c7db1f086084e8a7c4b14f481c6a117f169762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:34:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"39786ccf0d9ea7aaf4b8c0648de5cfc66ee24e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 21:34:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a71b0fa566e48cd98779075a30d3462\n\n\nWarning:\n  Dependency cycle detected and project openstack/ironic-lib doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":2},{"id":"45234a4c97e21534d40029dc47fab26d78cbccf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:44:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"586ec5f2e4bd3b90e593a9764cb23950d805c87d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 21:44:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f08781e79774f05a037ca0cd83fde4d\n\n\nWarning:\n  Dependency cycle detected and project openstack/ironic-lib doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":3},{"id":"74869e87395ec7eb353dd766f29b1e8706f641a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:47:53.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c0f33766491fa68134c47744289821b7f20194c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 21:48:18.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/project-config/zuul.d/projects.yaml@master\", line 2032, column 19","accounts_in_message":[],"_revision_number":4},{"id":"8446f15939da827e5a2d54ae3ac61798a2a5e619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:49:43.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4e2dfc9fa5bb99e112cf78c1841aed58566ebe63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 21:50:06.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  project: null\n  name: openstack/ironic-lib\n  templates:\n  - official-openstack-repo-jobs\n  ...","accounts_in_message":[],"_revision_number":5},{"id":"d8418132433b257bb0a883403b03b40e86e4944e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:55:11.000000000","message":"Topic set to ironic-lib-deprecation","accounts_in_message":[],"_revision_number":5},{"id":"d4ff1ffafc638e86176fcc31bec03c6fc2f66afd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 21:55:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3796d516f7fa1d7306c2fa41a868893bd6007667","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-14 21:58:35.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bc0c218cb5d13598036554aa91e4fa98bf85968b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-14 22:01:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"493115f3392a38734fadf45e8cc4a761243e46b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 22:02:18.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/939278 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"da12267d8b678cbb3f8b93043bba52240f3f8efd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-14 22:04:38.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThis looks good enough to start. Pay extra close attention the next time you request a stable point release to make sure it actually runs jobs. Zuul\u0027s handling of tag-to-branch relationships has changed over time (Git still considers tags to be rightly independent of any branch context since they can realistically appear on 0 or more branches in a repo), so I\u0027m hesitant to speculate on what will actually transpire. OpenStack deprecates repos so infrequently that, unless there\u0027s a recent working example, the best we can do is try and see what happens.","accounts_in_message":[],"_revision_number":6},{"id":"db4de3b12c80b30e4594a05ba16905499c1e5ee0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-14 22:06:08.000000000","message":"Patch Set 6: -Workflow\n\nOops, re-approve once the dependency has merged.","accounts_in_message":[],"_revision_number":6},{"id":"d5da0f7bc657c2f70c5dc1cde838a142bf983e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 22:09:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7496d1bbc22d4b3180dbf22fd410f67f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b14e319cd8574c678b7d4d72d626af08 : SUCCESS in 4m 11s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7c7f8e382047436cbf9fbafcec5f061a : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":6},{"id":"8ca9bc12cc1e7519e70dde8f5d427288d2dee5b4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-15 18:19:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e5b829df774278ff1d15ce42023e5bfd0d111a51","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-15 20:50:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7a14610d51e7c25360dbcf19a4add6752aa93faf","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-16 00:30:10.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"afdaeab612a3fea7883397f312870c605130ee86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-16 22:18:30.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5dd46cc3e67d807ccf3070f32b38637445257574","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-16 22:19:25.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b3613d5234e7b1aecce24653b0cb987a13b6885c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 22:20:26.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/939278 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"eeb6fa88d0a09f4c66e2e5ae888e2366ba60469f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-16 22:20:32.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"5ca685a80edfcba908564b9362a2a1dc2dac04c6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-17 01:59:21.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"07306a231eb7e48619a297b8a02f1cd571bec001","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-17 10:09:55.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"9b25d4835c1defa151ce2fb05c09eaafb50ddef0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-17 14:10:43.000000000","message":"Patch Set 7: Workflow+1\n\nTreating Ghanshyam\u0027s +1 as a +2.","accounts_in_message":[],"_revision_number":7},{"id":"157921e02d122bea5fd088d58e94153ba6cfa3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 14:45:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/def5d175de6048699e99efd34c6c9efd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d8de0913737424cb1e86917f47a6277 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":7},{"id":"af7666f60bc970dbc1f7c854e82f150eb3a13940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 14:46:37.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffbb8a5a6d084d38aeefe0b9dc659486\n\n\nWarning:\n  Change 939278 in project openstack/governance does not share a change queue with 939282 in project openstack/project-config","accounts_in_message":[],"_revision_number":7},{"id":"1bf989308dfb9f94a11a418f736294d3ab3548ab","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-17 15:42:56.000000000","message":"Patch Set 7: -Workflow\n\nrecheck (sorry, I missed that this Depends-On a WIP governance change)","accounts_in_message":[],"_revision_number":7},{"id":"9b204e0e30fc23744959bf77f46428028d673787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 15:54:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed76e3792377463dbb87003e428a226c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dbafcd80c4b4474399f82271d676ee4a : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":7},{"id":"082d152cfd7a1b5b87fa4550d7f9a16b9d43c6a9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-24 19:09:06.000000000","message":"Patch Set 7: Workflow+1\n\nDependency has merged now.","accounts_in_message":[],"_revision_number":7},{"id":"c9a637d2a0a823781d2b3384dd98f65d469531fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:10:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5e28758aea0ac5c5fde01195d6acbf15e6c8fcbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:16:18.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a6db1a96fbc4604ab9fb5e4fe6f3ddd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c77d6aa6ae342d6b1feb25a642c6786 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":7},{"id":"cba843ebe1a5a6bada6fe72694a70bfa93f3fdfa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:16:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"60ec88f0a75a60d203e69e6e2f82758361aa5c08","revisions":{"b3736f579cb4a7e6fae5cdfb33c4c94c86bb0e60":{"kind":"REWORK","_number":1,"created":"2025-01-14 21:25:33.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/82/939282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/939282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/939282/1"}}},"commit":{"parents":[{"commit":"7012a566681f7702da46bdea6acf3459febbda09","subject":"Repair promote-openstack-manuals-developer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7012a566681f7702da46bdea6acf3459febbda09"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"subject":"Deprecate ironic-lib","message":"Deprecate ironic-lib\n\nIronic-lib functionality has been moved, as needed, into Ironic and\nIronic-Python-Agent.\n\nDepends-on: https://review.opendev.org/c/openstack/ironic-lib/+/939277\nChange-Id: If073f91089d19ae0ad8815f6b6e5049e711bbd0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3736f579cb4a7e6fae5cdfb33c4c94c86bb0e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3736f579cb4a7e6fae5cdfb33c4c94c86bb0e60"}]},"branch":"refs/heads/master"},"906114eef95d26d8cfac7de5891f9ee180fd0c97":{"kind":"REWORK","_number":2,"created":"2025-01-14 21:34:28.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/82/939282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/939282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/939282/2"}}},"commit":{"parents":[{"commit":"7012a566681f7702da46bdea6acf3459febbda09","subject":"Repair promote-openstack-manuals-developer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7012a566681f7702da46bdea6acf3459febbda09"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:33:57.000000000","tz":-480},"subject":"Deprecate ironic-lib","message":"Deprecate ironic-lib\n\nIronic-lib functionality has been moved, as needed, into Ironic and\nIronic-Python-Agent.\n\nRelease jobs are being kept for active stable branches to ensure we can\nrelease should we have urgent fixes for stable branches until they age\nout of support.\n\nDepends-on: https://review.opendev.org/c/openstack/ironic-lib/+/939277\nChange-Id: If073f91089d19ae0ad8815f6b6e5049e711bbd0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/906114eef95d26d8cfac7de5891f9ee180fd0c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/906114eef95d26d8cfac7de5891f9ee180fd0c97"}]},"branch":"refs/heads/master"},"99242ddf6ac1112e62f36299f60e32399babf9f9":{"kind":"REWORK","_number":3,"created":"2025-01-14 21:44:08.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/82/939282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/939282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/939282/3"}}},"commit":{"parents":[{"commit":"7012a566681f7702da46bdea6acf3459febbda09","subject":"Repair promote-openstack-manuals-developer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7012a566681f7702da46bdea6acf3459febbda09"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:43:53.000000000","tz":-480},"subject":"Deprecate ironic-lib","message":"Deprecate ironic-lib\n\nIronic-lib functionality has been moved, as needed, into Ironic and\nIronic-Python-Agent.\n\nRelease jobs are being kept for active stable branches to ensure we can\nrelease should we have urgent fixes for stable branches until they age\nout of support.\n\nDepends-on: https://review.opendev.org/c/openstack/ironic-lib/+/939277\nChange-Id: If073f91089d19ae0ad8815f6b6e5049e711bbd0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/99242ddf6ac1112e62f36299f60e32399babf9f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/99242ddf6ac1112e62f36299f60e32399babf9f9"}]},"branch":"refs/heads/master"},"054a249b74a1cae303dab6b1e5cdec481c73deeb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-01-14 21:47:53.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/82/939282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/939282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/939282/4"}}},"commit":{"parents":[{"commit":"7012a566681f7702da46bdea6acf3459febbda09","subject":"Repair promote-openstack-manuals-developer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7012a566681f7702da46bdea6acf3459febbda09"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:47:53.000000000","tz":0},"subject":"Deprecate ironic-lib","message":"Deprecate ironic-lib\n\nIronic-lib functionality has been moved, as needed, into Ironic and\nIronic-Python-Agent.\n\nRelease jobs are being kept for active stable branches to ensure we can\nrelease should we have urgent fixes for stable branches until they age\nout of support.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/939278\nChange-Id: If073f91089d19ae0ad8815f6b6e5049e711bbd0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/054a249b74a1cae303dab6b1e5cdec481c73deeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/054a249b74a1cae303dab6b1e5cdec481c73deeb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"64451bd104a441d456fca750d3de0209c04ebb88":{"kind":"REWORK","_number":5,"created":"2025-01-14 21:49:43.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/82/939282/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/939282/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/939282/5"}}},"commit":{"parents":[{"commit":"7012a566681f7702da46bdea6acf3459febbda09","subject":"Repair promote-openstack-manuals-developer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7012a566681f7702da46bdea6acf3459febbda09"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:49:43.000000000","tz":0},"subject":"Deprecate ironic-lib","message":"Deprecate ironic-lib\n\nIronic-lib functionality has been moved, as needed, into Ironic and\nIronic-Python-Agent.\n\nRelease jobs are being kept for active stable branches to ensure we can\nrelease should we have urgent fixes for stable branches until they age\nout of support.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/939278\nChange-Id: If073f91089d19ae0ad8815f6b6e5049e711bbd0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64451bd104a441d456fca750d3de0209c04ebb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64451bd104a441d456fca750d3de0209c04ebb88"}]},"branch":"refs/heads/master"},"8b3fec2a307c2277500e3d951928b45511ea14a0":{"kind":"REWORK","_number":6,"created":"2025-01-14 21:55:52.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/82/939282/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/939282/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/939282/6"}}},"commit":{"parents":[{"commit":"7012a566681f7702da46bdea6acf3459febbda09","subject":"Repair promote-openstack-manuals-developer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7012a566681f7702da46bdea6acf3459febbda09"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:54:43.000000000","tz":-480},"subject":"Deprecate ironic-lib","message":"Deprecate ironic-lib\n\nIronic-lib functionality has been moved, as needed, into Ironic and\nIronic-Python-Agent.\n\nRelease jobs are being kept for active stable branches to ensure we can\nrelease should we have urgent fixes for stable branches until they age\nout of support.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/939278\nChange-Id: If073f91089d19ae0ad8815f6b6e5049e711bbd0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b3fec2a307c2277500e3d951928b45511ea14a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b3fec2a307c2277500e3d951928b45511ea14a0"}]},"branch":"refs/heads/master"},"60ec88f0a75a60d203e69e6e2f82758361aa5c08":{"kind":"REWORK","_number":7,"created":"2025-01-16 22:18:30.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/82/939282/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/939282/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/939282/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/939282/7"}}},"commit":{"parents":[{"commit":"7012a566681f7702da46bdea6acf3459febbda09","subject":"Repair promote-openstack-manuals-developer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7012a566681f7702da46bdea6acf3459febbda09"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-14 21:25:03.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2025-01-16 22:18:25.000000000","tz":-480},"subject":"Deprecate ironic-lib","message":"Deprecate ironic-lib\n\nIronic-lib functionality has been moved, as needed, into Ironic and\nIronic-Python-Agent.\n\nRelease jobs are being kept for active stable branches to ensure we can\nrelease should we have urgent fixes for stable branches until they age\nout of support.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/939278\nChange-Id: If073f91089d19ae0ad8815f6b6e5049e711bbd0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60ec88f0a75a60d203e69e6e2f82758361aa5c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60ec88f0a75a60d203e69e6e2f82758361aa5c08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
