)]}'
{"id":"openstack%2Fproject-config~940845","triplet_id":"openstack%2Fproject-config~master~I448d889eaafaddc2271737d2df8e82647ef55dcd","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-02-07 18:01:32.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-02-07 18:01:32.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-02-07 18:01:32.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-02-07 18:01:32.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-02-07 17:57:22.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I448d889eaafaddc2271737d2df8e82647ef55dcd","subject":"Add separate ACL for openstackdocstheme","status":"MERGED","created":"2025-02-06 03:13:36.000000000","updated":"2025-02-07 18:51:07.000000000","submitted":"2025-02-07 18:01:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940845","meta_rev_id":"be67f469d534168f282d41f2fd7c5be878822328","_number":940845,"virtual_id_number":940845,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-07 18:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":2,"date":"2025-02-07 17:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2025-02-07 17:57:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-06 03:16:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-02-06 03:16:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-02-06 03:17:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-02-06 03:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-06 03:33:19.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-02-07 14:23:47.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"27fb762e1727ce24018f9c8be38f913d7c6eb3c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-06 03:13:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad9618b64e8c402f8e7c45319a72ef10605547f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 03:22:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07e032a5e9304b12ac7cf15a43e4d5ab\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8a3e4e4350642e195e6b99e5d38ff48 : SUCCESS in 6m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3fd183951f9946d68b16c432fa3c2663 : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4fae2f7cb4bd95d979f0dfb777bca561408839f2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-06 03:33:19.000000000","message":"Patch Set 1:\n\nIs the backport stuff relevant? That repo does\u0027t (currently) have stable branches at all.","accounts_in_message":[],"_revision_number":1},{"id":"e80afa847b429892c1674bde793b3fa6d1e88e5c","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-06 05:45:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c888cce768edf2e951fab8f02ab628bed01f7823","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-07 02:46:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bad867cdeaf44422e186a7b279a73e068e7d6d2d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-07 02:57:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8c9377c7c931dc7aa9c8522e9b77de5e782cd3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-07 02:59:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"418b0e96f4096c69738657d27d24f93a417cb80d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-07 03:02:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e404c70929257c846d5205976d03ad1be73147f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 03:04:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3257d604d29a4467bf37665c99256405\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/079239fb97d743ee9e30a14846cb7d74 : SUCCESS in 4m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ea6b774cdd2d4819b07d1bf4fba3de35 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"2bf2162bbc382e1c3c4b6a4286ffbb13f64be42e","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-07 07:00:19.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a966f0846ff55b852373ec3c65b28f50cf312f6a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-07 14:23:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nSee inline comment.","accounts_in_message":[],"_revision_number":2},{"id":"ad91d2dba93da09b643f962a6296a0f934b55d2d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-07 17:21:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fe00e8ca74c6cf64ffd1a3acf959523698a0ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-07 17:21:51.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"968a58306856491b70a1421adc187287303321c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-07 17:23:12.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a9ce721bc6d20caba282e520afa4b8b5c40b96e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 17:30:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d43d82fbb74f52a8b9144f8ceb6265\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7aca1d394bcf4b8e8f6b82761ae35f87 : SUCCESS in 6m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d34e783242c740f6ba7d5dea8dbc0fcc : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"1d3470f7eed02f6985d97f8a41cd2586583815f5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-07 17:57:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5d9e5c3c95ab33363c13c38d90a79fff1ab50e7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 17:57:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0087f019282e17aec48fce5af74698bb2958fb68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:01:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/236c13838d214d1f950ded207ecdfb7b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/50fabb96f82c401d866d5df82fd9138b : SUCCESS in 3m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f23eb1d3fdc442f4905b7fd712d3698a : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":4},{"id":"df716a202724b184a97dbba50b228dd574809a80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:01:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8f7cd4045ef1fdc121e93262ea54da9213c74c90","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:22:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/034070efd9a84757b38ecb23c5fef782\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f5937aa52e224178bb112a2e058cf9f7 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"be67f469d534168f282d41f2fd7c5be878822328","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-07 18:51:07.000000000","message":"Patch Set 4:\n\nI\u0027ve included oslo-core into the new openstackdocstheme-core now.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"87f636c115b3abcf092cbf18e06bd6db66108aa0","revisions":{"da1338b683b691c46a2ed8c9e443bb33f00ee1c5":{"kind":"REWORK","_number":1,"created":"2025-02-06 03:13:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/940845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/940845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/940845/1"}}},"commit":{"parents":[{"commit":"3b59458dfecb39c42bd312ade7cf1c830489b64a","subject":"let core fix WIP state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b59458dfecb39c42bd312ade7cf1c830489b64a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-06 03:10:23.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-06 03:10:23.000000000","tz":-480},"subject":"Add separate ACL for openstackdocstheme","message":"Add separate ACL for openstackdocstheme\n\nAs discussed in oslo meeting[1], we are adding separate\nACL for openstackdocstheme deliverabes. Initially, oslo-core[2]\ncan be added as \u0027Included Groups\u0027 and after that oslo team can\nadd more core member.\n\n[1] https://etherpad.opendev.org/p/epoxy-oslo-meeting-tracking#L93\n[2] https://review.opendev.org/admin/groups/oslo-core\n\nChange-Id: I448d889eaafaddc2271737d2df8e82647ef55dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da1338b683b691c46a2ed8c9e443bb33f00ee1c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da1338b683b691c46a2ed8c9e443bb33f00ee1c5"}]},"branch":"refs/heads/master"},"369269c8a06bfaf03b770c9083914e6150899f68":{"kind":"REWORK","_number":2,"created":"2025-02-07 02:59:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/940845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/940845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/940845/2"}}},"commit":{"parents":[{"commit":"3b59458dfecb39c42bd312ade7cf1c830489b64a","subject":"let core fix WIP state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b59458dfecb39c42bd312ade7cf1c830489b64a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-06 03:10:23.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-07 02:59:25.000000000","tz":-480},"subject":"Add separate ACL for openstackdocstheme","message":"Add separate ACL for openstackdocstheme\n\nAs discussed in oslo meeting[1], we are adding separate\nACL for openstackdocstheme deliverabes. Initially, oslo-core[2]\ncan be added as \u0027Included Groups\u0027 and after that oslo team can\nadd more core member.\n\n[1] https://etherpad.opendev.org/p/epoxy-oslo-meeting-tracking#L93\n[2] https://review.opendev.org/admin/groups/oslo-core\n\nChange-Id: I448d889eaafaddc2271737d2df8e82647ef55dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/369269c8a06bfaf03b770c9083914e6150899f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/369269c8a06bfaf03b770c9083914e6150899f68"}]},"branch":"refs/heads/master"},"ad7d21be281c36558f132b372e3434a660162bac":{"kind":"REWORK","_number":3,"created":"2025-02-07 17:21:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/940845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/940845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/940845/3"}}},"commit":{"parents":[{"commit":"3b59458dfecb39c42bd312ade7cf1c830489b64a","subject":"let core fix WIP state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b59458dfecb39c42bd312ade7cf1c830489b64a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-06 03:10:23.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2025-02-07 17:21:51.000000000","tz":0},"subject":"Add separate ACL for openstackdocstheme","message":"Add separate ACL for openstackdocstheme\n\nAs discussed in oslo meeting[1], we are adding separate\nACL for openstackdocstheme deliverabes. Initially, oslo-core[2]\ncan be added as \u0027Included Groups\u0027 and after that oslo team can\nadd more core member.\n\n[1] https://etherpad.opendev.org/p/epoxy-oslo-meeting-tracking#L93\n[2] https://review.opendev.org/admin/groups/oslo-core\n\nChange-Id: I448d889eaafaddc2271737d2df8e82647ef55dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad7d21be281c36558f132b372e3434a660162bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad7d21be281c36558f132b372e3434a660162bac"}]},"branch":"refs/heads/master"},"87f636c115b3abcf092cbf18e06bd6db66108aa0":{"kind":"REWORK","_number":4,"created":"2025-02-07 17:23:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/940845/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/940845/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/940845/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/940845/4"}}},"commit":{"parents":[{"commit":"3b59458dfecb39c42bd312ade7cf1c830489b64a","subject":"let core fix WIP state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b59458dfecb39c42bd312ade7cf1c830489b64a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-06 03:10:23.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2025-02-07 17:23:12.000000000","tz":0},"subject":"Add separate ACL for openstackdocstheme","message":"Add separate ACL for openstackdocstheme\n\nAs discussed in oslo meeting[1], we are adding separate\nACL for openstackdocstheme deliverabes. Initially, oslo-core[2]\ncan be added as \u0027Included Groups\u0027 and after that oslo team can\nadd more core member.\n\n[1] https://etherpad.opendev.org/p/epoxy-oslo-meeting-tracking#L93\n[2] https://review.opendev.org/admin/groups/oslo-core\n\nChange-Id: I448d889eaafaddc2271737d2df8e82647ef55dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/87f636c115b3abcf092cbf18e06bd6db66108aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/87f636c115b3abcf092cbf18e06bd6db66108aa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
