)]}'
{"id":"openstack%2Fproject-config~943106","triplet_id":"openstack%2Fproject-config~master~Icaea80853e6c8ecb8951d7cacd69d6db3853b3ce","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-03-06 15:44:28.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-03-06 15:46:33.000000000","reason":"Change was submitted"}},"hashtags":["flex-dfw3"],"change_id":"Icaea80853e6c8ecb8951d7cacd69d6db3853b3ce","subject":"Boot nodes in Rackspace Flex SJC3 again","status":"MERGED","created":"2025-03-01 17:29:20.000000000","updated":"2025-03-06 15:55:13.000000000","submitted":"2025-03-06 15:46:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"943106","meta_rev_id":"a1b874990ab71df90839738c6b9c8cf33a2087e4","_number":943106,"virtual_id_number":943106,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 15:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-03-06 15:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-06 15:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-06 15:22:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-01 17:37:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-03 17:11:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"c55dbf26034e18dbb9f706a9e3875649fb77e12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-01 17:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51b5674783f94c44ae3440d5934c801d56e29abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 17:37:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0927f82b132c411eb146dbdabac8af93\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85dd69ef0e584cd79d260fda8f9629ed : SUCCESS in 4m 22s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2370c514f5dc4f5c96f686e823877dd7 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"55882f21bd418f88f5179a57c1b25ab9f2efc4ad","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-03 16:57:56.000000000","message":"Hashtag added: flex-dfw3","accounts_in_message":[],"_revision_number":1},{"id":"12c6f1141e05154957680e0b1f2fa4149e73bed1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-03 17:11:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0709d65589d325f9a42b3322a9b67f3ab02c4eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-06 15:01:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1f7d4d4b692efa6d0d8bc8e10337ec13965fcb0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:18:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436aaf4b95be4ecd9b77948e4eee8374\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/25fa34d7f8f24966942570912f784ddb : SUCCESS in 4m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/39fff84d9f134736bfa1e6d62ad969c2 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6b81caa5135682e847683686485c7206db4a6cf5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-06 15:22:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nSelf-approving to get this moving, Clark offered a +2 in IRC but isn\u0027t in a position to relay it to Gerrit at the moment.","accounts_in_message":[],"_revision_number":2},{"id":"cebf5224827e56805b4141a6b3f6ef4ca8fcbc4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:22:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db27cb0639c8c8f0dc8894faacf79373a92e2738","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-06 15:44:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"17f8cd474ba315f8f7081bd8027a1119c4e4a83c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:46:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcdf05733cd6453b847de6ba9cdc896c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/86777b70653146e79fc683d0a85931ad : SUCCESS in 4m 17s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4783118dbf3947c38ef3d09e0c3e84b1 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7d0a7a763377ba33b51c838589d81b3ddd93f7fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:46:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a1b874990ab71df90839738c6b9c8cf33a2087e4","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:55:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d3fddb55fd24abaaa3778fca7a5d5e8\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/cc776cf015684553906072b195f217f3 : SUCCESS in 8m 20s\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/23106499877b44d3848da4138f33c63f : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33cc33a2610c2016562d5836e9cbb774bdc1e2c7","revisions":{"cf0d40d28134f6256a97b9584698e4a0e48267b4":{"kind":"REWORK","_number":1,"created":"2025-03-01 17:29:20.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/06/943106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/943106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/943106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/943106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/943106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/943106/1"}}},"commit":{"parents":[{"commit":"1e236908aacabc16e1e4250bf72a85b6457d74cb","subject":"Add the DFW3 region for Rackspace Flex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e236908aacabc16e1e4250bf72a85b6457d74cb"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-03-01 17:27:35.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-03-01 17:27:35.000000000","tz":0},"subject":"Boot nodes in both Rackspace Flex regions","message":"Boot nodes in both Rackspace Flex regions\n\nNow that there are images available in DFW3 and SJC3, set\nmax-servers for them to the value we previously used in the old SJC3\nproject. We can try increasing these later once we see how this\ngoes.\n\nChange-Id: Icaea80853e6c8ecb8951d7cacd69d6db3853b3ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cf0d40d28134f6256a97b9584698e4a0e48267b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cf0d40d28134f6256a97b9584698e4a0e48267b4"}]},"branch":"refs/heads/master"},"33cc33a2610c2016562d5836e9cbb774bdc1e2c7":{"kind":"REWORK","_number":2,"created":"2025-03-06 15:01:40.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/06/943106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/943106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/943106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/943106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/943106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/943106/2"}}},"commit":{"parents":[{"commit":"1e236908aacabc16e1e4250bf72a85b6457d74cb","subject":"Add the DFW3 region for Rackspace Flex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e236908aacabc16e1e4250bf72a85b6457d74cb"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-03-01 17:27:35.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-03-06 15:00:24.000000000","tz":0},"subject":"Boot nodes in Rackspace Flex SJC3 again","message":"Boot nodes in Rackspace Flex SJC3 again\n\nNow that there are images available in SJC3, set max-servers to the\nsame value we previously used in our old project there, as that\u0027s\nstill the maximum we can fit in the current RAM quota.\n\nChange-Id: Icaea80853e6c8ecb8951d7cacd69d6db3853b3ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/33cc33a2610c2016562d5836e9cbb774bdc1e2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/33cc33a2610c2016562d5836e9cbb774bdc1e2c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
