)]}'
{"id":"openstack%2Fproject-config~943647","triplet_id":"openstack%2Fproject-config~master~I6849da93cfe788729e654b95685ff576f0811993","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-03-06 22:33:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6849da93cfe788729e654b95685ff576f0811993","subject":"Correct flavor name for Rackspace Flex DFW3","status":"MERGED","created":"2025-03-06 22:07:49.000000000","updated":"2025-03-06 22:42:03.000000000","submitted":"2025-03-06 22:33:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943647","meta_rev_id":"873470a376a87d76316dc00a4edaf4d0fe1900dd","_number":943647,"virtual_id_number":943647,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 22:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-03-06 22:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-03-06 22:11:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 22:11:38.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-03-06 22:19:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad1a824d0d19fb39218150d5af3628b73e0f03da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-06 22:07:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ff5c2f4470728a94b650fca8ad5a4a9c56a0cb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-06 22:11:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1125b3064da8c2277a017e05392bbe46338e4599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:19:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7f889c46c444749f05e2d670e9e9ae\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f0146dd30474eb1ab1874d9094061b7 : SUCCESS in 4m 34s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d3e275a7534141fdb4133714134e5555 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ffa244aa2ed6e4f0c8606461f72212ea4c8c2b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:19:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86f96c6932cabebd0d9fbeffd2edad31ef4faeab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:33:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c10bb0ca9a640e5a3dae260877694a8\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/64478c535cac42d88b68b0b6da85eadf : SUCCESS in 5m 49s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/49396c2264944a87a170ef323d0960cf : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8dbff7fc0697e4b9b456c60a61fd8a49f0181479","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:33:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"873470a376a87d76316dc00a4edaf4d0fe1900dd","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:42:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f9b4dbe7b0d438c9623e729a6455c9b\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/bc3d94293edd4fa98cb47d221f8e347f : SUCCESS in 8m 14s\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/78f19b054bd44471b5cda2b989e7f381 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7b1fc2b588a52080901bbff6493760482c562c5","revisions":{"b7b1fc2b588a52080901bbff6493760482c562c5":{"kind":"REWORK","_number":1,"created":"2025-03-06 22:07:49.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/47/943647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/943647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/943647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/943647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/943647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/943647/1"}}},"commit":{"parents":[{"commit":"4c59cf1bf2a709634a3549f7e609b82ecc831951","subject":"Start booting nodes in Rackspace Flex DFW3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c59cf1bf2a709634a3549f7e609b82ecc831951"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-03-06 22:05:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-03-06 22:05:15.000000000","tz":0},"subject":"Correct flavor name for Rackspace Flex DFW3","message":"Correct flavor name for Rackspace Flex DFW3\n\nThe flavors in DFW3 are not the same as in SJC3. For the latter\nwe\u0027ve been using gp.0.4.8 but in the former the equivalent is\ngp.5.4.8 instead. Fix it.\n\nChange-Id: I6849da93cfe788729e654b95685ff576f0811993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7b1fc2b588a52080901bbff6493760482c562c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7b1fc2b588a52080901bbff6493760482c562c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
