)]}'
{"id":"openstack%2Fproject-config~945359","triplet_id":"openstack%2Fproject-config~master~I22d35939e9b5ff996f57a3053d43cfc46aae025f","project":"openstack/project-config","branch":"master","topic":"new-nodepool-launchers","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-03-24 18:02:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I22d35939e9b5ff996f57a3053d43cfc46aae025f","subject":"Add config files for nl05, nl06, nl07 and nl08","status":"MERGED","created":"2025-03-24 16:33:35.000000000","updated":"2025-03-24 18:15:02.000000000","submitted":"2025-03-24 18:02:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945359-new-nodepool-launchers","meta_rev_id":"7f51e9328b2f69b6bf6930388e89b9e4c4b3c505","_number":945359,"virtual_id_number":945359,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 18:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-24 17:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-24 17:57:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 16:45:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-24 17:34:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"ccb150841d5d3c935a160beb5fa0961a32f41941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-24 16:33:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e137e60b124fd135e6c926a524c2fef5f47a46f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:45:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e67b296cf6c949cfab8ab2a86a5da055\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d9dfa09843344bafa124c6f79b54f7f1 : SUCCESS in 2m 59s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/03d48cab415742279f9fc29342829e09 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"60369c5e2b18aa3fa6736b98bf52173166308c20","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-24 17:34:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12816d3f0676d499b65d2cb0d017349b5bd5ccff","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-24 17:57:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"810f45b2b7434036fdf228d7656b0ae41590cd88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 17:57:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f655c54e2421b2138570a1b92147c4557c5dc5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:02:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e10197b53204d9eb024c8a090151776\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0401ed2c365e46bfb61ecfedd766c039 : SUCCESS in 2m 46s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e700711489334d2c857f42acf52169fb : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"004276c2f08d9c8fee9d11a9b378e13a14676f84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:02:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7f51e9328b2f69b6bf6930388e89b9e4c4b3c505","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:15:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9524cfe83fe4429a60312f66783a4ce\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/5b773ab811984e12a78068128a1b2aa5 : SUCCESS in 6m 06s\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/51ab827fdf494506bef29c7e5ed7ee28 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0cb346a09b30793337ca03dd20468b7e66e7ed8","revisions":{"d0cb346a09b30793337ca03dd20468b7e66e7ed8":{"kind":"REWORK","_number":1,"created":"2025-03-24 16:33:35.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/59/945359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/945359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/945359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/945359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/945359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/945359/1"}}},"commit":{"parents":[{"commit":"2aecd9b962d48cfe3f8fabdbff2d469bb296d52c","subject":"Cleanup nb04 builder config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2aecd9b962d48cfe3f8fabdbff2d469bb296d52c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-03-24 16:29:54.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-03-24 16:29:54.000000000","tz":-420},"subject":"Add config files for nl05, nl06, nl07 and nl08","message":"Add config files for nl05, nl06, nl07 and nl08\n\nnl05, nl06, nl07, nl08 are four new Noble nodepool launcher nodes and\nthey need configs. We give them configs by making direct copies of the\nold nl01, nl02, nl03, and nl04 configs. nl01 is copied to nl05, nl02 is\ncopied to nl06 and so on. The only edits I have made are to update the\ncomments indicating nl01 is responsible for managing min ready nodes.\nThat responsibility is now nl05\u0027s and the comment is updated to reflect\nthat in the new config file.\n\nTo roll this out my plan is to land this change. Then update the\ninventory to add nl06. As soon as that inventory update change begins to\ndeploy nl02 will go in the emergency file and I will manually shutdown\nits launcher. Then nl06 should come up and begin serving the providers\nthat were running on nl02. I have chosen nl02/nl06 as the first pair as\nimpact is minimal (only openmetal). Depending on how this goes I may\ncontinue to do one launcher at a time of batch up the remaining three.\n\nChange-Id: I22d35939e9b5ff996f57a3053d43cfc46aae025f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d0cb346a09b30793337ca03dd20468b7e66e7ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d0cb346a09b30793337ca03dd20468b7e66e7ed8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
