)]}'
{"id":"openstack%2Fproject-config~945715","triplet_id":"openstack%2Fproject-config~master~I8a82c035f3d5fffb9e233cac2b66faaa5c295518","project":"openstack/project-config","branch":"master","topic":"add-viommu","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-03-27 17:24:23.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-03-27 17:38:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a82c035f3d5fffb9e233cac2b66faaa5c295518","subject":"Add viommu support to vexxhost ubuntu-noble","status":"MERGED","created":"2025-03-27 15:00:14.000000000","updated":"2025-03-27 17:45:55.000000000","submitted":"2025-03-27 17:38:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945715-add-viommu","meta_rev_id":"722583d7c424e3aa29453feee7094d204ed4deb5","_number":945715,"virtual_id_number":945715,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 17:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-03-27 17:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-27 17:29:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-27 17:29:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-27 15:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-27 15:36:49.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-03-27 17:29:00.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"349676415b0c019851b9d1779938f4e43b9d6b64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-03-27 15:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2f5cefe6c414d0b99adac29938ce8e9ec4ec204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:12:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71b3db12d2b34fcdacf2b657ba8b68bd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c11efefaf5ef4bf19ae6c8875a805ab7 : SUCCESS in 4m 51s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/6854d4b892274337aa24f373fca6ab85 : FAILURE in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6452d445d367b792acdd614e13cf55f4cb45d80a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-27 15:36:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24830d7db63a84eb49ba1ed24063ab95b2cc9d3b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-03-27 15:38:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b2d3aa8beb93298bdef97a3a0c837b6d6f1473d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-03-27 15:38:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e4f8846bd54f2d980249019eddc1dd4e43971416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:47:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144debd71793467b9cc12122446f6a09\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8374d1e237e4ae2a9db3b88ffef7a52 : SUCCESS in 3m 39s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b6751c38f6714bd9ac8deae5e2695ceb : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"006a15e688a17605055542600898d4eca19a874e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-27 15:55:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6994332ebf10df4a490e6997c46b8bc38e0c6e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-03-27 17:22:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5cd31c0e6ee52690b6cbcb3f497a28081d1cda88","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-27 17:24:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3dddba85db1570306392b0e049c63b1dda5f6f4d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-27 17:29:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"95f15350aacd4dc5bacad945d5e4b669863a60b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:33:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540379e9e7fb43e0bb97abcea0466764\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9fb79456bdbc4df787edbbb8aaff9656 : SUCCESS in 5m 00s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b8b5682a539b4e05b028a3d18d4304ea : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3035899cd9bccabd9f7c6d0a619d860b703227bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:33:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fefab27819a5b9624e808eb2975712338586156a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:38:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f300d7720cf4912b4c6dd87df50352f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/287a8b2af4af426dafd7636024c5e0ba : SUCCESS in 4m 15s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/43297feda19e4b598b590dcb7e1ed247 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":3},{"id":"581e69a840ab539883792042727ccdab8674f333","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:38:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"722583d7c424e3aa29453feee7094d204ed4deb5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:45:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29964bbb51ac4d8ba9eec283231078f7\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/7d92c9fb536349ecba01dba08b078ee0 : SUCCESS in 6m 38s\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/2646c580eae14ffa8df9cd035055a71b : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5d8760a39a45b0228f0df66df0e0ede77a1be6ec","revisions":{"5730ad0bd9927692145f7556c4345ada4d767ef4":{"kind":"REWORK","_number":1,"created":"2025-03-27 15:00:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/945715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/945715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/945715/1"}}},"commit":{"parents":[{"commit":"7b63533b629ec294ab1d3a60242ac65e57d0838b","subject":"Update SSL use in our IRC access check script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b63533b629ec294ab1d3a60242ac65e57d0838b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-27 14:59:14.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-27 14:59:14.000000000","tz":-420},"subject":"Add viommu support to vexxhost ubuntu-noble","message":"Add viommu support to vexxhost ubuntu-noble\n\nSince we have other labels for weird configurations (like nested virt)\nhere, it makes sense to also add viommu support to the noble instances\non vexxhost so we can test things like PCI passthrough.\n\nChange-Id: I8a82c035f3d5fffb9e233cac2b66faaa5c295518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5730ad0bd9927692145f7556c4345ada4d767ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5730ad0bd9927692145f7556c4345ada4d767ef4"}]},"branch":"refs/heads/master"},"c909f7b99c42dba10ff5916827349df920fe7130":{"kind":"REWORK","_number":2,"created":"2025-03-27 15:38:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/945715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/945715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/945715/2"}}},"commit":{"parents":[{"commit":"7b63533b629ec294ab1d3a60242ac65e57d0838b","subject":"Update SSL use in our IRC access check script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b63533b629ec294ab1d3a60242ac65e57d0838b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-27 14:59:14.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-27 15:38:18.000000000","tz":-420},"subject":"Add viommu support to vexxhost ubuntu-noble","message":"Add viommu support to vexxhost ubuntu-noble\n\nSince we have other labels for weird configurations (like nested virt)\nhere, it makes sense to also add viommu support to the noble instances\non vexxhost so we can test things like PCI passthrough.\n\nChange-Id: I8a82c035f3d5fffb9e233cac2b66faaa5c295518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c909f7b99c42dba10ff5916827349df920fe7130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c909f7b99c42dba10ff5916827349df920fe7130"}]},"branch":"refs/heads/master"},"5d8760a39a45b0228f0df66df0e0ede77a1be6ec":{"kind":"REWORK","_number":3,"created":"2025-03-27 17:22:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/945715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/945715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/945715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/945715/3"}}},"commit":{"parents":[{"commit":"7b63533b629ec294ab1d3a60242ac65e57d0838b","subject":"Update SSL use in our IRC access check script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b63533b629ec294ab1d3a60242ac65e57d0838b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-27 14:59:14.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-03-27 17:22:52.000000000","tz":-420},"subject":"Add viommu support to vexxhost ubuntu-noble","message":"Add viommu support to vexxhost ubuntu-noble\n\nSince we have other labels for weird configurations (like nested virt)\nhere, it makes sense to also add viommu support to the noble instances\non vexxhost so we can test things like PCI passthrough.\n\nChange-Id: I8a82c035f3d5fffb9e233cac2b66faaa5c295518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d8760a39a45b0228f0df66df0e0ede77a1be6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d8760a39a45b0228f0df66df0e0ede77a1be6ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
