)]}'
{"id":"openstack%2Fproject-config~946742","triplet_id":"openstack%2Fproject-config~master~Ic09d95dc35b1f8ffb50b56e9a14e888d961aeea5","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-04-10 12:34:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic09d95dc35b1f8ffb50b56e9a14e888d961aeea5","subject":"Add jobs for new Aetos repository","status":"MERGED","created":"2025-04-09 08:17:53.000000000","updated":"2025-04-10 12:37:13.000000000","submitted":"2025-04-10 12:34:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946742","meta_rev_id":"ae4b977c1fbde60be168a48412ae90687bafa79d","_number":946742,"virtual_id_number":946742,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-10 12:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-10 12:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-10 12:31:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-09 08:18:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-10 12:31:08.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"9d7cd84fb144f7af347fc65253051da06de21f29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-04-09 08:17:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bef43396fa5f4f06d32a0c8f35856ae6167f38e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 08:18:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/aetos\" was not found.  All projects referenced\n  within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/aetos\" project stanza:\n\n  project:\n      name: openstack/aetos\n      templates:\n        - official-openstack-repo-jobs\n        - translation-jobs-master-only\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 565","accounts_in_message":[],"_revision_number":1},{"id":"54b883fcafa84895ff98dae729135f44cd93c2c5","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-04-10 07:23:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc0379b137c95afd20ac6ae144e1f164f9b2298f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 07:23:29.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/project-config for 946742,1","accounts_in_message":[],"_revision_number":1},{"id":"8fcd3c9047089c753f1bc4040624daf2b13b182f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-04-10 07:25:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a24c4d68bb752afa5f5370baeb275085019a09a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 07:33:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85a9ab9492d94c46ba4fc8fe661b6d4f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/baa6934f1a72493cba65ea3b5f12044d : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f29b1d37185c405a6011d780e913516c15c6dab1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-10 12:31:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f15e05795b67ddad60f990cefccf2478e728683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 12:31:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ec2b121497ff537c22ce717aaf6cf87b3066013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 12:34:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78db8127a24d4f0093541e528e094099\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2e6a2e8b15034e10aca8d34ebcc340b1 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1ef217e757300fd9644d640a07f928433c5e2caa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 12:34:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ae4b977c1fbde60be168a48412ae90687bafa79d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 12:37:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b8d08aa48f84a15be522a6db32c87fa\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/785b16f5ac274ba280095882ec976c09 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3276fd31f52a9fe1177a926ff2d70fc26c27cc01","revisions":{"bf315b4c22ed08b89aeb5e7e9acd5e8bcfded958":{"kind":"REWORK","_number":1,"created":"2025-04-09 08:17:53.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/42/946742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/946742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/946742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/946742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/946742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/946742/1"}}},"commit":{"parents":[{"commit":"bdcd134e3ad6520a669888df4aebf6777bfe6c2e","subject":"Add Aetos under the telemetry project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bdcd134e3ad6520a669888df4aebf6777bfe6c2e"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-04-09 07:51:05.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-04-09 07:51:05.000000000","tz":-240},"subject":"Add jobs for new Aetos repository","message":"Add jobs for new Aetos repository\n\nChange-Id: Ic09d95dc35b1f8ffb50b56e9a14e888d961aeea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf315b4c22ed08b89aeb5e7e9acd5e8bcfded958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf315b4c22ed08b89aeb5e7e9acd5e8bcfded958"}]},"branch":"refs/heads/master"},"3276fd31f52a9fe1177a926ff2d70fc26c27cc01":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-10 07:25:22.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/42/946742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/946742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/946742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/946742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/946742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/946742/2"}}},"commit":{"parents":[{"commit":"ccded3fbdfbf664fbacc70372c4ae92e7fffa7ae","subject":"Add Aetos under the telemetry project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccded3fbdfbf664fbacc70372c4ae92e7fffa7ae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-04-09 07:51:05.000000000","tz":-240},"committer":{"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","date":"2025-04-10 07:25:22.000000000","tz":0},"subject":"Add jobs for new Aetos repository","message":"Add jobs for new Aetos repository\n\nChange-Id: Ic09d95dc35b1f8ffb50b56e9a14e888d961aeea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3276fd31f52a9fe1177a926ff2d70fc26c27cc01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3276fd31f52a9fe1177a926ff2d70fc26c27cc01"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
