)]}'
{"id":"openstack%2Fproject-config~946767","triplet_id":"openstack%2Fproject-config~master~Idf0848c27b042a9caf3b0a53d5cf4fcbf7f0aa6e","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-04-21 23:00:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf0848c27b042a9caf3b0a53d5cf4fcbf7f0aa6e","subject":"Remove telemetry groups","status":"MERGED","created":"2025-04-09 14:02:04.000000000","updated":"2025-04-21 23:31:22.000000000","submitted":"2025-04-21 23:00:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946767","meta_rev_id":"5a053603d836ad8c4d05eb6a869a01fc48aaa7a5","_number":946767,"virtual_id_number":946767,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-21 23:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-04-09 16:16:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-21 22:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-21 22:50:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-09 14:10:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-09 16:16:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-04-21 22:50:05.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"e78ec72e7bf03c8d7cba79a9c6d6c97336286401","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-04-09 14:02:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87fa4b7fb12c156e53663f26e5cecaeabca5d748","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-04-09 14:04:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c56baff8de7d66a775539f3bcaf7d0205e11a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 14:10:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4879b1b161cd431dad1644b1b3c785af\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e41d9c92ced847f2b9fe7d24e36f0b4a : SUCCESS in 4m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/664a016643a748f896de6cb3c4dde4dc : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"60918f0fb2d8079f9d8a501b145432897ebc3875","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-09 16:16:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"252055da7b2dc48d8c701329e99bf268d00469f9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-21 22:50:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b96ceadb27da047ec2a86318415914fcf769abde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 22:50:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf50c27f6c0b1002ec4511c36e9f21ec0e662331","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:00:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef9e733807354d2096fd13a2aa4501d5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/41ffe9f7ec6449918dd49dafe95430be : SUCCESS in 4m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5537410807cf493798d2d4eda0f0e36c : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e7fbf66de66373c14e83d1c6ba01626550163e4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:00:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a053603d836ad8c4d05eb6a869a01fc48aaa7a5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14e882bd08b47d495a0496ade70aa68\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/4ef4b7f50dc64c36b2592d9607ea8a0f : SUCCESS in 9m 36s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/1c5e2ac0066e437d95a1e261f1520209 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd094fd200793c1ef37d41b061428a3fd7b0e0dc","revisions":{"cd094fd200793c1ef37d41b061428a3fd7b0e0dc":{"kind":"REWORK","_number":1,"created":"2025-04-09 14:02:04.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/67/946767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/67/946767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/946767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/946767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/946767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/67/946767/1"}}},"commit":{"parents":[{"commit":"ccded3fbdfbf664fbacc70372c4ae92e7fffa7ae","subject":"Add Aetos under the telemetry project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccded3fbdfbf664fbacc70372c4ae92e7fffa7ae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-04-09 13:50:02.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-04-09 13:50:02.000000000","tz":-240},"subject":"Remove telemetry groups","message":"Remove telemetry groups\n\nThe telemetry groups are a remnant from when the telemetry\nproject used storyboard. Telemetry uses launchpad now. These\nfields cause bug reporting to be misconfigured.\n\nChange-Id: Idf0848c27b042a9caf3b0a53d5cf4fcbf7f0aa6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd094fd200793c1ef37d41b061428a3fd7b0e0dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd094fd200793c1ef37d41b061428a3fd7b0e0dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
