)]}'
{"id":"openstack%2Fproject-config~947629","triplet_id":"openstack%2Fproject-config~master~I16d58a30bca0115f53b95bab2a23e619f8d1f4b3","project":"openstack/project-config","branch":"master","topic":"osa/deprecate_tests","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-04-21 23:08:36.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-04-18 12:45:30.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I16d58a30bca0115f53b95bab2a23e619f8d1f4b3","subject":"Deprecate openstack-ansible-tests repository","status":"MERGED","created":"2025-04-17 19:37:55.000000000","updated":"2025-04-21 23:56:21.000000000","submitted":"2025-04-21 23:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947629-osa/deprecate_tests","meta_rev_id":"fe439d59463a72a8e176720768844e4f3dec3d87","_number":947629,"virtual_id_number":947629,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-21 23:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2025-04-18 12:53:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-18 12:45:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-21 23:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-17 19:38:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-17 19:59:33.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-04-18 12:53:47.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"85e83fd0ca8f1b48e387c626f580b17fab0cfc09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-17 19:37:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a7a1fff58fb17c58cc2990f6c18333e74ecfaf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 19:38:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project template \"noop\" was not found.\n\nThe problem appears in the the \"openstack/openstack-ansible-tests\" project stanza:\n\n  project:\n      name: openstack/openstack-ansible-tests\n      templates:\n        - official-openstack-repo-jobs\n        - noop\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3085","accounts_in_message":[],"_revision_number":1},{"id":"768eec8fd227dca1673ac41d62e51329df865ba5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-17 19:59:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e16c68a66341e97c3c194d37c1ea9af6babaa221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-18 06:24:32.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6eeaa4a50d3fa2dfee810ea8024771649844540a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 06:29:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd46299dfdc4fa39cd958d4aa05c760\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d56f9a53e51c49b99faf96b62eec53dc : SUCCESS in 3m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1f48d05b8a984596a4530c6ada4728ed : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"2fd765f612145687032646f3d0bd25cc1f1faa62","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-18 12:45:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1761d8bb81f62aa2ba03af98b31931cc203ec16f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-04-18 12:53:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bdccf039ca3b870c2b9ac605f5c20c5313ae9dcc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-18 13:15:10.000000000","message":"Patch Set 2: -Workflow\n\nOh, right, the parent has a not-yet-merged Depends-On to a project in a separate gate queue.","accounts_in_message":[],"_revision_number":2},{"id":"4b12c75cdbfb9387e922773a7c5de69c9add48e4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-21 23:00:51.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b28f1241cb876112f4291e1a7fdb2eb61cc14e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:01:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b71eaef543d90a32b221f41fb4b79511640fcbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:08:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebbff1d88ed04e8b88749dba78adf9e6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/697c98e414b14a88b9676a7e346f362b : SUCCESS in 5m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1ae2064dde5f4d39aafe8b71175a98f7 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d98b2d3324f524912551f20ee97a7703684e8e5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:08:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fe439d59463a72a8e176720768844e4f3dec3d87","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:56:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08290bfcf936499a8e4f96e440ad2783\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/e53fd943a9f142bbb9de36d2f8802d6e : SUCCESS in 4m 56s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/8f604c2ce436462781866558ec5c6b27 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea2edaf971045378d2ec58e8e89825ee19a2f1b2","revisions":{"ea36757edf2638f8239411b99e024630cb7d6e60":{"kind":"REWORK","_number":1,"created":"2025-04-17 19:37:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/947629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/947629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/947629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/947629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/947629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/947629/1"}}},"commit":{"parents":[{"commit":"ff9aab67ffcd1d86cb196d0d4509bcee26ab51d2","subject":"Move OSA sync to integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff9aab67ffcd1d86cb196d0d4509bcee26ab51d2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-17 19:37:45.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-17 19:37:45.000000000","tz":120},"subject":"Deprecate openstack-ansible-tests repository","message":"Deprecate openstack-ansible-tests repository\n\nAs part of the deprecation process we mark repository as deprecated\nalong with adding noop jobs to proceed with content removal\n\nChange-Id: I16d58a30bca0115f53b95bab2a23e619f8d1f4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ea36757edf2638f8239411b99e024630cb7d6e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ea36757edf2638f8239411b99e024630cb7d6e60"}]},"branch":"refs/heads/master"},"ea2edaf971045378d2ec58e8e89825ee19a2f1b2":{"kind":"REWORK","_number":2,"created":"2025-04-18 06:24:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/947629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/947629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/947629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/947629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/947629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/947629/2"}}},"commit":{"parents":[{"commit":"ff9aab67ffcd1d86cb196d0d4509bcee26ab51d2","subject":"Move OSA sync to integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff9aab67ffcd1d86cb196d0d4509bcee26ab51d2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-17 19:37:45.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-04-18 06:24:31.000000000","tz":0},"subject":"Deprecate openstack-ansible-tests repository","message":"Deprecate openstack-ansible-tests repository\n\nAs part of the deprecation process we mark repository as deprecated\nalong with adding noop jobs to proceed with content removal\n\nChange-Id: I16d58a30bca0115f53b95bab2a23e619f8d1f4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ea2edaf971045378d2ec58e8e89825ee19a2f1b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ea2edaf971045378d2ec58e8e89825ee19a2f1b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
