)]}'
{"id":"openstack%2Fproject-config~948249","triplet_id":"openstack%2Fproject-config~master~Id7c152a01a8561baae2c188fde9cb9b80a2c2bcf","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-04-28 18:00:15.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-04-28 10:51:50.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-04-28 17:53:20.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id7c152a01a8561baae2c188fde9cb9b80a2c2bcf","subject":"create openstack/grian-ui git repo","status":"MERGED","created":"2025-04-25 16:20:01.000000000","updated":"2025-04-28 18:11:31.000000000","submitted":"2025-04-28 18:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"948249-new-project","meta_rev_id":"113d63f6f8ab965dd084442070f991c7c9920e38","_number":948249,"virtual_id_number":948249,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-28 18:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-04-28 10:51:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-28 17:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-28 17:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-25 16:28:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-25 16:29:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-04-28 10:42:19.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-04-28 17:53:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"5dee1cde2f08d48cd15820cc7153e3247908ffb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 16:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37480391e08ccfcf0eec7747935c70dbb578cd4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 16:28:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e43d763f6b7244d8ab25ce358a5505fd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/97a56536e58d4d8980049d021a2b2434 : SUCCESS in 4m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/344f0fe59ade4764bb63793e4fb92c8d : FAILURE in 3m 02s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/73c2d3df48cc48ccb89711fe33fb0ae0 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4488f0a84904507869f3b26139e8b7fb173e31a3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-25 16:29:29.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nRemember to propose an openstack/governance change for this as well. Ideally, Juan should +1 either this or the governance change as Telemetry PTL before we approve creation of the repository.","accounts_in_message":[],"_revision_number":1},{"id":"9563cefb6c8d78c2d7535f264ffbb25406a2b881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 16:31:29.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3d898e98f737ccb0dbfe18ce4ca5a9347fdaf3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 16:38:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d002babba8c43b497da68725950379b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7217c153a0204fe792a209c163415ade : SUCCESS in 2m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4bf5fd60e59943a4888bd631ebe64a6f : FAILURE in 2m 44s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4b32cf0b58fa446297db3325f85d53f9 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4a675fa5a5c151d4300ff7050c85e58a5b4e8b1b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 16:55:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"741fb573b410cd8520b1a33fd2c524ebf6c4db66","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 16:56:49.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00daf4c2b7a60c952eab04b68cbfba56df7c2fa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-26 21:42:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dd49ed9e9d1dd612471468879dc7dd3d6e9d838f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-26 21:47:05.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9a7ca1b38a29c3fc3d302096e019c04e51b6a89f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 21:52:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65ec5d83b3b94be99192101aa5e847ba\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a68771faa864c569c60fec3491b8d02 : SUCCESS in 3m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/08daac1c5ee0467d9aa5e37d95ef7b6e : FAILURE in 1m 49s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ca3829ecb3954d3e88c126656baea634 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"3269dd7e8d5eea7d0537c7d099a145f3d633d565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-28 10:36:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5ebb60f00c9c4576cf93b8ab0b7759f803385ec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 10:45:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8729b78f92b43fea0205118b078512a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8534d981a1d4187902340d54fa55880 : SUCCESS in 4m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7222f049893040f8b8c4f12b29849883 : SUCCESS in 3m 41s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/23f1ff9c0e0b4d11b649200761cab401 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"eef0085e6e41bb8bf5ad682448a702f2fe84829f","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-04-28 10:51:50.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9f899f2f1ac4222182f8157390408f0ae9a1455c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-28 17:44:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"782114e3233e14741e70be4324d33d7f41f5b016","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-04-28 17:53:20.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"48282ef03f38977fd66dcde544a2201324855508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 17:53:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a90220693dea6d4e405a621bfb3a127fb78176c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 18:00:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54bb3cb579b042f0882184b4dcd290f0\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6cdd08811dc84bdab523186a58fb12d9 : SUCCESS in 4m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fd4c09cd3aa74aa384a163e8c2e2c8f3 : SUCCESS in 3m 46s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/59918e610015405998b840c1f15f7675 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":5},{"id":"d74b086beac12c4fa6838a20c525893d263835f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 18:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"113d63f6f8ab965dd084442070f991c7c9920e38","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 18:11:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b545d57e6b4e49a32bd64f23b3e5e7\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/2f273ce675344f448f9f9a190b5b4b07 : SUCCESS in 11m 04s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/56835335c14d42e8b4279345c78c9874 : SUCCESS in 3m 37s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/f12fb9b856b44612bd8f58e7d163d3a0 : SUCCESS in 5m 49s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/bd8fd24186c3406daa26231d662ddfef : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4bc39d7e7625e5d3de9f292b2a01ae156de1448a","revisions":{"ce9801f5323028fa08b518c0bac02911e90ae159":{"kind":"REWORK","_number":1,"created":"2025-04-25 16:20:01.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/948249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/948249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/948249/1"}}},"commit":{"parents":[{"commit":"51824e32dfc11adcaf7d19848b1536bc18bc6622","subject":"Merge \"Remove noop jobs for openstack-ansible-tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51824e32dfc11adcaf7d19848b1536bc18bc6622"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 15:53:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 15:53:59.000000000","tz":60},"subject":"create openstack/grian-ui git repo","message":"create openstack/grian-ui git repo\n\ngrian is the irish word for sun\nThe grian-ui repo will host a new horizon\nplugin to \"enlighten\" the dashboard with metrics.\n\nThis will be an offical deliverable of the Telemetry\nteam. While an inital prototype exists the repo\nwill be started from scratch so there is no upstream\nrepo to import.\n\nThis was discussed with both the horizon and telemetry\nteams during the recent ptg.\n\nhttps: //etherpad.opendev.org/p/r.23126f9476afd06ca072ad95ec096a0d#L158\nhttps: //etherpad.opendev.org/p/r.72ac6a7268e4b9d854f75715adede80c#L33\nChange-Id: Id7c152a01a8561baae2c188fde9cb9b80a2c2bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce9801f5323028fa08b518c0bac02911e90ae159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce9801f5323028fa08b518c0bac02911e90ae159"}]},"branch":"refs/heads/master"},"bc83e20ea75d980e67b7023f5d02d1b0eac27e9e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-25 16:31:29.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/948249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/948249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/948249/2"}}},"commit":{"parents":[{"commit":"51824e32dfc11adcaf7d19848b1536bc18bc6622","subject":"Merge \"Remove noop jobs for openstack-ansible-tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51824e32dfc11adcaf7d19848b1536bc18bc6622"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 15:53:59.000000000","tz":60},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2025-04-25 16:31:29.000000000","tz":0},"subject":"create openstack/grian-ui git repo","message":"create openstack/grian-ui git repo\n\ngrian is the irish word for sun\nThe grian-ui repo will host a new horizon\nplugin to \"enlighten\" the dashboard with metrics.\n\nThis will be an offical deliverable of the Telemetry\nteam. While an inital prototype exists the repo\nwill be started from scratch so there is no upstream\nrepo to import.\n\nThis was discussed with both the horizon and telemetry\nteams during the recent ptg.\n\nhttps: //etherpad.opendev.org/p/r.23126f9476afd06ca072ad95ec096a0d#L158\nhttps: //etherpad.opendev.org/p/r.72ac6a7268e4b9d854f75715adede80c#L33\nChange-Id: Id7c152a01a8561baae2c188fde9cb9b80a2c2bcf\nNeeded-By: I37b49d588e682e8dba62468c4df3c02265b0f305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc83e20ea75d980e67b7023f5d02d1b0eac27e9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc83e20ea75d980e67b7023f5d02d1b0eac27e9e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f04113cae30ccb552dc5b3785534ece35fab6f4b":{"kind":"REWORK","_number":3,"created":"2025-04-26 21:42:49.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/948249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/948249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/948249/3"}}},"commit":{"parents":[{"commit":"51824e32dfc11adcaf7d19848b1536bc18bc6622","subject":"Merge \"Remove noop jobs for openstack-ansible-tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51824e32dfc11adcaf7d19848b1536bc18bc6622"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 15:53:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-26 21:37:54.000000000","tz":60},"subject":"create openstack/grian-ui git repo","message":"create openstack/grian-ui git repo\n\ngrian is the irish word for sun\nThe grian-ui repo will host a new horizon\nplugin to \"enlighten\" the dashboard with metrics.\n\nThis will be an offical deliverable of the Telemetry\nteam. While an inital prototype exists the repo\nwill be started from scratch so there is no upstream\nrepo to import.\n\nThis was discussed with both the horizon and telemetry\nteams during the recent ptg.\n\nhttps: //etherpad.opendev.org/p/r.23126f9476afd06ca072ad95ec096a0d#L158\nhttps: //etherpad.opendev.org/p/r.72ac6a7268e4b9d854f75715adede80c#L33\nChange-Id: Id7c152a01a8561baae2c188fde9cb9b80a2c2bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f04113cae30ccb552dc5b3785534ece35fab6f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f04113cae30ccb552dc5b3785534ece35fab6f4b"}]},"branch":"refs/heads/master"},"f2b944a0196a602161f02d14560d08bbd613d7e4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-26 21:47:05.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/948249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/948249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/948249/4"}}},"commit":{"parents":[{"commit":"51824e32dfc11adcaf7d19848b1536bc18bc6622","subject":"Merge \"Remove noop jobs for openstack-ansible-tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51824e32dfc11adcaf7d19848b1536bc18bc6622"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 15:53:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-26 21:45:52.000000000","tz":60},"subject":"create openstack/grian-ui git repo","message":"create openstack/grian-ui git repo\n\ngrian is the irish word for sun\nThe grian-ui repo will host a new horizon\nplugin to \"enlighten\" the dashboard with metrics.\n\nThis will be an offical deliverable of the Telemetry\nteam. While an inital prototype exists the repo\nwill be started from scratch so there is no upstream\nrepo to import.\n\nThis was discussed with both the horizon and telemetry\nteams during the recent ptg.\n\nhttps: //etherpad.opendev.org/p/r.23126f9476afd06ca072ad95ec096a0d#L158\nhttps: //etherpad.opendev.org/p/r.72ac6a7268e4b9d854f75715adede80c#L33\nNeeded-By: I37b49d588e682e8dba62468c4df3c02265b0f305\nChange-Id: Id7c152a01a8561baae2c188fde9cb9b80a2c2bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2b944a0196a602161f02d14560d08bbd613d7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2b944a0196a602161f02d14560d08bbd613d7e4"}]},"branch":"refs/heads/master"},"4bc39d7e7625e5d3de9f292b2a01ae156de1448a":{"kind":"REWORK","_number":5,"created":"2025-04-28 10:36:11.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/948249/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/948249/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/948249/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/948249/5"}}},"commit":{"parents":[{"commit":"51824e32dfc11adcaf7d19848b1536bc18bc6622","subject":"Merge \"Remove noop jobs for openstack-ansible-tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51824e32dfc11adcaf7d19848b1536bc18bc6622"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 15:53:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-28 10:35:12.000000000","tz":60},"subject":"create openstack/grian-ui git repo","message":"create openstack/grian-ui git repo\n\ngrian is the irish word for sun\nThe grian-ui repo will host a new horizon\nplugin to \"enlighten\" the dashboard with metrics.\n\nThis will be an offical deliverable of the Telemetry\nteam. While an inital prototype exists the repo\nwill be started from scratch so there is no upstream\nrepo to import.\n\nThis was discussed with both the horizon and telemetry\nteams during the recent ptg.\n\nhttps: //etherpad.opendev.org/p/r.23126f9476afd06ca072ad95ec096a0d#L158\nhttps: //etherpad.opendev.org/p/r.72ac6a7268e4b9d854f75715adede80c#L33\nNeeded-By: I37b49d588e682e8dba62468c4df3c02265b0f305\nChange-Id: Id7c152a01a8561baae2c188fde9cb9b80a2c2bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39d7e7625e5d3de9f292b2a01ae156de1448a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39d7e7625e5d3de9f292b2a01ae156de1448a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
