)]}'
{"id":"openstack%2Fproject-config~948250","triplet_id":"openstack%2Fproject-config~master~I82bb6b3d0299e7eadfb376afd51564dce8a7ecee","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-08-06 17:52:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82bb6b3d0299e7eadfb376afd51564dce8a7ecee","subject":"Add jobs for openstack/grian-ui repo","status":"MERGED","created":"2025-04-25 16:20:01.000000000","updated":"2025-08-06 17:55:29.000000000","submitted":"2025-08-06 17:52:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948250-new-project","meta_rev_id":"95eaccc1fea9b318c84da90545a44274f8e0041a","_number":948250,"virtual_id_number":948250,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 17:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 17:45:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-06 17:45:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-25 16:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-06 17:45:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"2eada49d7fd421c082c629cd990a8ff9e252231f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 16:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8b41084d7061ed21a0654d08ba286a98ee5e16a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 16:20:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/grian-ui\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/grian-ui\" project stanza:\n\n  project:\n      name: openstack/grian-ui\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1931","accounts_in_message":[],"_revision_number":1},{"id":"f672d3b0fe2e3ccea4677318c99b82e5b7343076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 16:32:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"940447552ee258a6d0df28b2a131f04f44c2250d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 16:32:58.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/grian-ui\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/grian-ui\" project stanza:\n\n  project:\n      name: openstack/grian-ui\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1931","accounts_in_message":[],"_revision_number":2},{"id":"7f26b062e0db64523ecd297892330aa1ca5bfab9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 17:05:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"34098a94ba0e925fc72bbd162341df7f81dd102e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 17:05:22.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/grian-ui\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/grian-ui\" project stanza:\n\n  project:\n      name: openstack/grian-ui\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1931","accounts_in_message":[],"_revision_number":3},{"id":"b08e5d606734119e81bfc27a27b3ae31e867b00a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 17:06:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4133fe23cfdcf995149135f8b06de78c8cfb40d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 17:06:28.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/grian-ui\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/grian-ui\" project stanza:\n\n  project:\n      name: openstack/grian-ui\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1931","accounts_in_message":[],"_revision_number":4},{"id":"b170ce8940df60c314a987ec2c03778d92047896","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 23:31:52.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"34becd38935dbcdf2fc7cf6dfc02dd30df26e2b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-26 21:42:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d9f9ce38b35651205259326414d2176d9f5baaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 21:43:11.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/grian-ui\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/grian-ui\" project stanza:\n\n  project:\n      name: openstack/grian-ui\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1935","accounts_in_message":[],"_revision_number":5},{"id":"5f68da688d358f6721f4aef14e25895d3986b64b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-26 21:47:05.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6a804b8123d22ed6badb8ebf772fd29e301b1198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 21:47:20.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/grian-ui\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/grian-ui\" project stanza:\n\n  project:\n      name: openstack/grian-ui\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1935","accounts_in_message":[],"_revision_number":6},{"id":"c9c6a6ef6e82eddd2b1afad7f1fac040e2b67a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-28 10:36:11.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"4778b8b824c274e6e1d16615a7c507b44459c372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 10:36:26.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/grian-ui\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/grian-ui\" project stanza:\n\n  project:\n      name: openstack/grian-ui\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1935","accounts_in_message":[],"_revision_number":7},{"id":"4868e0489887d71e6f49d4d4724b9da4936a1477","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-28 18:11:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5f6157503fade276dabff57d75503a3b344e8d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 18:36:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc98d0863894ab197989326cec4d96e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/844bce2613024bd5ba83fa7f99e86c44 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":7},{"id":"af7b78da1d86669761a0694b42c9fa838c248641","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-30 17:38:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"51e4976173fece9acc19073cf3581f5c6e835d93","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-08-06 17:45:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7fdb5c2c292d566de9a8f3c289297a1e909905b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:46:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"509e96059a4f92220b2f35c930361a87e6605951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:52:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96152a65a160478ea20cae62f33ea812\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/120273e00b6e48cb89726f63dd733a0f : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":7},{"id":"33fc4949ce1fc981785f9ab6ed3b502f358443e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:52:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"95eaccc1fea9b318c84da90545a44274f8e0041a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:55:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92962bb8149d4adaa4a54a14b8f2aabd\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/283d9fd66a7c4cd28c47112aef221d31 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"50e0d7081caf8a12cba5446bc6f0dffa0cea0921","revisions":{"66c3cd3daad4ff55931ab3127868a8fe409ad862":{"kind":"REWORK","_number":1,"created":"2025-04-25 16:20:01.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/50/948250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/948250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/948250/1"}}},"commit":{"parents":[{"commit":"ce9801f5323028fa08b518c0bac02911e90ae159","subject":"create openstack/grian-ui git repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce9801f5323028fa08b518c0bac02911e90ae159"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:16:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:18:56.000000000","tz":60},"subject":"Add jobs for openstack/grian-ui repo","message":"Add jobs for openstack/grian-ui repo\n\nChange-Id: I82bb6b3d0299e7eadfb376afd51564dce8a7ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66c3cd3daad4ff55931ab3127868a8fe409ad862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66c3cd3daad4ff55931ab3127868a8fe409ad862"}]},"branch":"refs/heads/master"},"47ed5b643356ac46fb35ba8c3edbc931c21b03dc":{"kind":"NO_CHANGE","_number":2,"created":"2025-04-25 16:32:34.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/50/948250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/948250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/948250/2"}}},"commit":{"parents":[{"commit":"bc83e20ea75d980e67b7023f5d02d1b0eac27e9e","subject":"create openstack/grian-ui git repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc83e20ea75d980e67b7023f5d02d1b0eac27e9e"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:16:59.000000000","tz":60},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2025-04-25 16:32:34.000000000","tz":0},"subject":"Add jobs for openstack/grian-ui repo","message":"Add jobs for openstack/grian-ui repo\n\nChange-Id: I82bb6b3d0299e7eadfb376afd51564dce8a7ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/47ed5b643356ac46fb35ba8c3edbc931c21b03dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/47ed5b643356ac46fb35ba8c3edbc931c21b03dc"}]},"branch":"refs/heads/master","description":"Rebase"},"217fbdad242524a38ddde083f4e5d1598e277295":{"kind":"REWORK","_number":3,"created":"2025-04-25 17:05:06.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/50/948250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/948250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/948250/3"}}},"commit":{"parents":[{"commit":"ce9801f5323028fa08b518c0bac02911e90ae159","subject":"create openstack/grian-ui git repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce9801f5323028fa08b518c0bac02911e90ae159"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:16:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 17:03:55.000000000","tz":60},"subject":"Add jobs for openstack/grian-ui repo","message":"Add jobs for openstack/grian-ui repo\n\nChange-Id: I82bb6b3d0299e7eadfb376afd51564dce8a7ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/217fbdad242524a38ddde083f4e5d1598e277295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/217fbdad242524a38ddde083f4e5d1598e277295"}]},"branch":"refs/heads/master"},"77d4ec509e00661646c5e3615e3c2f6b4ac91d99":{"kind":"NO_CHANGE","_number":4,"created":"2025-04-25 17:06:11.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/50/948250/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/948250/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/948250/4"}}},"commit":{"parents":[{"commit":"bc83e20ea75d980e67b7023f5d02d1b0eac27e9e","subject":"create openstack/grian-ui git repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc83e20ea75d980e67b7023f5d02d1b0eac27e9e"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:16:59.000000000","tz":60},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2025-04-25 17:06:11.000000000","tz":0},"subject":"Add jobs for openstack/grian-ui repo","message":"Add jobs for openstack/grian-ui repo\n\nChange-Id: I82bb6b3d0299e7eadfb376afd51564dce8a7ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/77d4ec509e00661646c5e3615e3c2f6b4ac91d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/77d4ec509e00661646c5e3615e3c2f6b4ac91d99"}]},"branch":"refs/heads/master","description":"Rebase"},"d8870d6ace62f8af1c7b7bc008d207da77bb2b25":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-04-26 21:42:49.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/50/948250/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/948250/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/948250/5"}}},"commit":{"parents":[{"commit":"f04113cae30ccb552dc5b3785534ece35fab6f4b","subject":"create openstack/grian-ui git repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f04113cae30ccb552dc5b3785534ece35fab6f4b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:16:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-26 21:37:55.000000000","tz":60},"subject":"Add jobs for openstack/grian-ui repo","message":"Add jobs for openstack/grian-ui repo\n\nChange-Id: I82bb6b3d0299e7eadfb376afd51564dce8a7ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8870d6ace62f8af1c7b7bc008d207da77bb2b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8870d6ace62f8af1c7b7bc008d207da77bb2b25"}]},"branch":"refs/heads/master"},"8e06760ff3dd67d0a9388c3079b3dc05acb99a4c":{"kind":"NO_CHANGE","_number":6,"created":"2025-04-26 21:47:05.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/50/948250/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/948250/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/948250/6"}}},"commit":{"parents":[{"commit":"f2b944a0196a602161f02d14560d08bbd613d7e4","subject":"create openstack/grian-ui git repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2b944a0196a602161f02d14560d08bbd613d7e4"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:16:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-26 21:46:53.000000000","tz":60},"subject":"Add jobs for openstack/grian-ui repo","message":"Add jobs for openstack/grian-ui repo\n\nChange-Id: I82bb6b3d0299e7eadfb376afd51564dce8a7ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e06760ff3dd67d0a9388c3079b3dc05acb99a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e06760ff3dd67d0a9388c3079b3dc05acb99a4c"}]},"branch":"refs/heads/master"},"50e0d7081caf8a12cba5446bc6f0dffa0cea0921":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-28 10:36:11.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/50/948250/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/948250/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/948250/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/948250/7"}}},"commit":{"parents":[{"commit":"4bc39d7e7625e5d3de9f292b2a01ae156de1448a","subject":"create openstack/grian-ui git repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39d7e7625e5d3de9f292b2a01ae156de1448a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-25 16:16:59.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-04-28 10:35:24.000000000","tz":60},"subject":"Add jobs for openstack/grian-ui repo","message":"Add jobs for openstack/grian-ui repo\n\nChange-Id: I82bb6b3d0299e7eadfb376afd51564dce8a7ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50e0d7081caf8a12cba5446bc6f0dffa0cea0921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50e0d7081caf8a12cba5446bc6f0dffa0cea0921"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
