)]}'
{"id":"openstack%2Fproject-config~949555","triplet_id":"openstack%2Fproject-config~master~I7e31987eb2256edae21089b67361ae895cc5986c","project":"openstack/project-config","branch":"master","topic":"qihuiz-2","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2025-05-14 13:29:37.000000000","reason":"Change was submitted"},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2025-05-13 03:39:45.000000000","reason":"\u003cGERRIT_ACCOUNT_35582\u003e replied on the change","reason_account":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"}},"35582":{"account":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"last_update":"2025-05-14 13:29:37.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-05-14 13:15:29.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7e31987eb2256edae21089b67361ae895cc5986c","subject":"to create a new repo for a cfn new launched sub-group heterogeneous distributed training framework","status":"MERGED","created":"2025-05-13 03:31:13.000000000","updated":"2025-05-15 02:41:00.000000000","submitted":"2025-05-14 13:29:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949555-qihuiz-2","meta_rev_id":"b3b8bb77703027b59017ae7696c67128d530ab8c","_number":949555,"virtual_id_number":949555,"owner":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-14 13:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-14 13:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-14 13:15:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 03:38:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-13 03:39:12.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-05-13 03:39:12.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2025-05-13 03:39:12.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2025-05-13 03:39:45.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REMOVED"},{"updated":"2025-05-13 03:39:45.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REMOVED"},{"updated":"2025-05-13 03:39:45.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REMOVED"},{"updated":"2025-05-13 03:47:52.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-05-13 03:47:52.000000000","updated_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"f5029305e346bb4082d0b0806ebfac863540d2a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-05-13 03:31:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb5addb23f232b6e857cf91418b1c481b8dde9f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-05-13 03:33:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3c16085ebb93dbbfcdf95226b105f4d56cd4e358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 03:38:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c98e213b6c84c2483907607b80c8c81\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d0b3fcda341e4d4c9c95b61507d739a1 : SUCCESS in 4m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fc7d703b29e54f258e7af37b1bf30a8f : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1a6fb88767a80a4c3bb7543815a2fc932eda7fd8","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-05-13 03:39:45.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_27623\u003e.","accounts_in_message":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"_revision_number":2},{"id":"b7facca497502696ea7c6beb0cdbc0cd14fca377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-05-13 03:42:24.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c1d1b6128c013875aba7fea46efc9c17a295a7f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 03:46:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5496fd29c7bf472f9ef10175a45d16a0\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/827dda0511474188b9482fc68ac40de5 : SUCCESS in 3m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2ce7723fe03b47f38b60e7e5316545d4 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"98a3e171602d352a219caf1ea596f53e553cd639","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-05-14 11:08:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1dfe581b3c8cc4cff45d2069f627d1697c083c18","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-05-14 13:15:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"76101a955b12a01dff16579fef6b6f59eec9f85e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:15:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fb7fb1f2c7368b05ec97a3610bfedd627171acf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:29:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/685a293fbdb14b8f9c7daff6a3f8694e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/23c79b01a76148c88e351e4e6736e43b : SUCCESS in 4m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b3128ecb558c4ad1b3999607fe45b2cf : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"8e12742b5f6184cb07de1b234378e8a6d1a9cd5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:29:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5b1c07da2100be570caaec50083477f35e267810","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:42:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/737ca4ed010c46e98ef6b2149a6b189e\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/67f95464a5934daa96fe37477a0ab873 : SUCCESS in 12m 31s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/b32c61c4f2f3469389b7362ea3e70c91 : SUCCESS in 3m 47s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/d0bcae297d124326a17a2185d0cdf5f4 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f25de6822bd7d499162f8ca7943b6d38d06df6a2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-05-14 14:38:13.000000000","message":"Patch Set 3:\n\nI\u0027ve added Qihui Zhao\u0027s account as the initial member of the heterogeneous-distributed-training-framework-core group.","accounts_in_message":[],"_revision_number":3},{"id":"b3b8bb77703027b59017ae7696c67128d530ab8c","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-05-15 02:41:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d7c91c54ef222e9965a5ab1ff90eb8579a4e6e16","revisions":{"edc3888210659b5bf45608068c249c20f0a410b2":{"kind":"REWORK","_number":1,"created":"2025-05-13 03:31:13.000000000","uploader":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"ref":"refs/changes/55/949555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/949555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/949555/1"}}},"commit":{"parents":[{"commit":"d709e82f709f83f21beb2ab851797357048cd2cf","subject":"Add zuul-launcher dashboards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d709e82f709f83f21beb2ab851797357048cd2cf"}]}],"author":{"name":"qihuiz93","email":"qihuiz93@163.com","date":"2025-05-13 03:30:22.000000000","tz":480},"committer":{"name":"qihuiz93","email":"qihuiz93@163.com","date":"2025-05-13 03:31:04.000000000","tz":480},"subject":"to create a new repo for a cfn new launched sub-group heterogeneous distributed training framework","message":"to create a new repo for a cfn new launched sub-group heterogeneous distributed training framework\n\nChange-Id: I7e31987eb2256edae21089b67361ae895cc5986c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edc3888210659b5bf45608068c249c20f0a410b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edc3888210659b5bf45608068c249c20f0a410b2"}]},"branch":"refs/heads/master"},"c86557825efed15500d07b72054f00feb9ec208c":{"kind":"REWORK","_number":2,"created":"2025-05-13 03:33:58.000000000","uploader":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"ref":"refs/changes/55/949555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/949555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/949555/2"}}},"commit":{"parents":[{"commit":"d709e82f709f83f21beb2ab851797357048cd2cf","subject":"Add zuul-launcher dashboards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d709e82f709f83f21beb2ab851797357048cd2cf"}]}],"author":{"name":"qihuiz93","email":"qihuiz93@163.com","date":"2025-05-13 03:30:22.000000000","tz":480},"committer":{"name":"Qihui Zhao","email":"qihuiz93@163.com","date":"2025-05-13 03:33:45.000000000","tz":0},"subject":"to create a new repo for a cfn new launched sub-group heterogeneous distributed training framework","message":"to create a new repo for a cfn new launched sub-group heterogeneous distributed training framework\n\nChange-Id: I7e31987eb2256edae21089b67361ae895cc5986c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c86557825efed15500d07b72054f00feb9ec208c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c86557825efed15500d07b72054f00feb9ec208c"}]},"branch":"refs/heads/master"},"d7c91c54ef222e9965a5ab1ff90eb8579a4e6e16":{"kind":"REWORK","_number":3,"created":"2025-05-13 03:42:24.000000000","uploader":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"ref":"refs/changes/55/949555/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/949555/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/949555/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/949555/3"}}},"commit":{"parents":[{"commit":"d709e82f709f83f21beb2ab851797357048cd2cf","subject":"Add zuul-launcher dashboards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d709e82f709f83f21beb2ab851797357048cd2cf"}]}],"author":{"name":"qihuiz93","email":"qihuiz93@163.com","date":"2025-05-13 03:30:22.000000000","tz":480},"committer":{"name":"Qihui Zhao","email":"qihuiz93@163.com","date":"2025-05-13 03:42:24.000000000","tz":0},"subject":"to create a new repo for a cfn new launched sub-group heterogeneous distributed training framework","message":"to create a new repo for a cfn new launched sub-group heterogeneous distributed training framework\n\nChange-Id: I7e31987eb2256edae21089b67361ae895cc5986c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7c91c54ef222e9965a5ab1ff90eb8579a4e6e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7c91c54ef222e9965a5ab1ff90eb8579a4e6e16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
