)]}'
{"id":"openstack%2Fproject-config~952547","triplet_id":"openstack%2Fproject-config~master~I430986e9ae4ff365738562b3b69d1d8982c6c904","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"38053":{"account":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"last_update":"2025-06-13 08:00:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I430986e9ae4ff365738562b3b69d1d8982c6c904","subject":"Enable release team to create branches and signed tags","status":"ABANDONED","created":"2025-06-13 06:18:35.000000000","updated":"2025-06-13 08:00:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a83fa360e47dcaf6801640725ef6a639c3093c5a","_number":952547,"virtual_id_number":952547,"owner":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-13 06:22:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-13 06:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23b0f7e73fe75d5ca79ebf6d8cba2f7697205280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-13 06:18:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8c6b185672f59737be1aa5a45defd0b6d51371b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 06:22:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3845737f394841ecad5eeef7af7eea7d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f2525509d6a347248b15d9a9a57ae67d : SUCCESS in 3m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7b5b210cc7e04e31b9e3109297f70133 : FAILURE in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a83fa360e47dcaf6801640725ef6a639c3093c5a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-13 08:00:25.000000000","message":"Abandoned\n\nBuild failed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6eea7841090ed2b08bdc0c2bfea9d8bd7712db7","revisions":{"d6eea7841090ed2b08bdc0c2bfea9d8bd7712db7":{"kind":"REWORK","_number":1,"created":"2025-06-13 06:18:35.000000000","uploader":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"ref":"refs/changes/47/952547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/952547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/952547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/952547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/952547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/952547/1"}}},"commit":{"parents":[{"commit":"ba18406f6066aa41a11d5f050632149a4b4af33b","subject":"Deduplicate OpenStack ACL content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba18406f6066aa41a11d5f050632149a4b4af33b"}]}],"author":{"name":"tianyutong","email":"tianyutong@sd.chinamobile.com","date":"2025-06-13 06:17:51.000000000","tz":480},"committer":{"name":"tianyutong","email":"tianyutong@sd.chinamobile.com","date":"2025-06-13 06:18:16.000000000","tz":480},"subject":"Enable release team to create branches and signed tags","message":"Enable release team to create branches and signed tags\n\nChange-Id: I430986e9ae4ff365738562b3b69d1d8982c6c904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6eea7841090ed2b08bdc0c2bfea9d8bd7712db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6eea7841090ed2b08bdc0c2bfea9d8bd7712db7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
