)]}'
{"id":"openstack%2Fproject-config~952643","triplet_id":"openstack%2Fproject-config~master~I8c9c325eb78dae714f13b4418c3decf639b300c0","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"38053":{"account":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"last_update":"2025-06-17 12:45:15.000000000","reason":"Change was submitted"},"35582":{"account":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"last_update":"2025-06-17 01:38:40.000000000","reason":"\u003cGERRIT_ACCOUNT_35582\u003e replied on the change","reason_account":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-06-16 13:00:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I8c9c325eb78dae714f13b4418c3decf639b300c0","subject":"Add release branch creation permission","status":"MERGED","created":"2025-06-16 00:56:40.000000000","updated":"2025-06-17 12:58:40.000000000","submitted":"2025-06-17 12:45:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952643","meta_rev_id":"837368c4346fbdd8798c627594bab6d662a0c5ec","_number":952643,"virtual_id_number":952643,"owner":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-17 12:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"all":[{"value":1,"date":"2025-06-16 01:58:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-17 09:03:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},{"value":2,"date":"2025-06-17 12:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},{"value":1,"date":"2025-06-17 12:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-16 01:00:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 02:00:24.000000000","updated_by":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-06-16 06:37:53.000000000","updated_by":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"reviewer":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"state":"REVIEWER"}],"messages":[{"id":"269cdb6276a4ee4bf7b4c5b9dfb51dd81334460b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-16 00:56:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff33f7d0b7312d4bf5be60443322cf564cc68994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 01:00:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70b44df5add46c28f33b767e5037174\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c4e4b9671609412b9e49b1f74df3b56a : SUCCESS in 3m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b9610becb6b94c25894957513936a04a : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3f47430398640e0838c8e97bf504f5abc0804735","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-16 01:58:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f1fd34992625fceb196c01bbaca9acab0e975f1f","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-16 06:35:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1841666c6b25bdb2c1a317adcb9c90e28fc61f1d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-16 13:00:13.000000000","message":"Patch Set 1:\n\nThe heterogeneous-distributed-training-framework-release group doesn\u0027t exist yet. This change will create it, but it will be initially empty until a Gerrit sysadmin adds an initial member (who can then add and remove other members as needed). Whose Gerrit account should be the initial member of this new group?","accounts_in_message":[],"_revision_number":1},{"id":"2147dcb09c7015e89fe844420c479fd759bc2b7c","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-06-17 01:38:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7574a1c527a14a3c64b2e19e60f47fe847ccbfb7","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-17 08:01:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a2ab7410e726da230bf00a92e1b1756d8a57d13","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-06-17 09:03:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f12f0d75cba98fdb0548e74addfbe76bc7e3268","author":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"},"date":"2025-06-17 09:04:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3da6807a71030a6fbabe1c035bb25df5b2230dad","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-17 12:29:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"362fff9be580d760519228df769e3b0e677bd811","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:29:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93f8f59c15c3ca7fc10b6049244ac123587f1758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:45:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c5d37946644579a254fcbc44d4be6c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ec67fd3e537345dc9dc69e6c8caba297 : SUCCESS in 4m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0d7439f048cf4efbb951a3e48a965283 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3df945ff179595e3ddc4fb36bf66435a86a655ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:45:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c077403014306898942edf9491802a435e1ac20f","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:50:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b56d7ca55f7431da5f637f87baf9529\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/da45964d5ce94c4cadbd069deb6cc060 : SUCCESS in 4m 58s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/51e225afafd7450b999b89e7235f2603 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"837368c4346fbdd8798c627594bab6d662a0c5ec","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-17 12:58:40.000000000","message":"Patch Set 1:\n\nI\u0027ve added tianyutong\u0027s account as the initial member of the new heterogeneous-distributed-training-framework-release group.\n\nNote that this change limits branch creation in the cfn/heterogeneous-distributed-training-framework repository to that new group, no longer allowing members of heterogeneous-distributed-training-framework-core to perform that function.\n\nAlso be aware that branch deletion is a separate permission which you would need to grant explicitly within this same ACL file also if you want to be able to remove any branches again later.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a6353a6f5899b801cddac1b10ce95a23649dba8","revisions":{"7a6353a6f5899b801cddac1b10ce95a23649dba8":{"kind":"REWORK","_number":1,"created":"2025-06-16 00:56:40.000000000","uploader":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"ref":"refs/changes/43/952643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/952643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/952643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/952643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/952643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/952643/1"}}},"commit":{"parents":[{"commit":"8c9151dbc224bd38d7fc98c69ba74b1ee0004ca3","subject":"Allow branch creation for heterogeneous-distributed-training-framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c9151dbc224bd38d7fc98c69ba74b1ee0004ca3"}]}],"author":{"name":"tianyutong","email":"tianyutong@sd.chinamobile.com","date":"2025-06-16 00:55:40.000000000","tz":480},"committer":{"name":"tianyutong","email":"tianyutong@sd.chinamobile.com","date":"2025-06-16 00:56:15.000000000","tz":480},"subject":"Add release branch creation permission","message":"Add release branch creation permission\n\nChange-Id: I8c9c325eb78dae714f13b4418c3decf639b300c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a6353a6f5899b801cddac1b10ce95a23649dba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a6353a6f5899b801cddac1b10ce95a23649dba8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35582,"name":"Qihui Zhao","email":"qihuiz93@163.com","username":"qihuiz93"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
