)]}'
{"id":"openstack%2Fproject-config~952712","triplet_id":"openstack%2Fproject-config~master~I37579d022578cf35a5d33ab4902b5f0696a32e0d","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-06-17 22:08:45.000000000","reason":"removed on reply"}},"hashtags":["opendev-niz"],"change_id":"I37579d022578cf35a5d33ab4902b5f0696a32e0d","subject":"Switch all Zuul tenants to use niz nodesets","status":"MERGED","created":"2025-06-16 20:13:24.000000000","updated":"2025-06-24 18:50:57.000000000","submitted":"2025-06-24 18:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952712","meta_rev_id":"0e3a5e68f6d1eed8eba6de68cdcfccd5a7a6a32b","_number":952712,"virtual_id_number":952712,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-24 18:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2025-06-24 17:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2025-06-16 20:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-16 20:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2025-06-24 17:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-16 20:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 20:24:59.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-06-16 20:44:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"9c0ad090e430dac8ff5e9cc0fe0b277fc0583194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-16 20:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fb9a9a34af3bbd583c6e1e0681fde43cb06d138","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-16 20:14:01.000000000","message":"Hashtag added: opendev-niz","accounts_in_message":[],"_revision_number":1},{"id":"380358f57f53c49b4573dd12bd6f00b2e213f45f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 20:22:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0979ff9fb36946b3bbf77fc9a3ffa910\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d2885a4c5a44818bc69be074deb58e2 : SUCCESS in 4m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/12fff89783134ab9bb61a5409ea2b7cc : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"36110dd298331c010b2631d9b82af25948a5c179","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-16 20:24:59.000000000","message":"Patch Set 1: Code-Review+2\n\nThis seems safe as a next step. Obviously any jobs that use anonymous nodesets will still be referencing Nodepool-supplied labels for the time being.","accounts_in_message":[],"_revision_number":1},{"id":"9d5e0b80f270c0400ba0ea5a07f7eb9380e02034","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-06-16 20:44:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4622c1e2a280e1040c64c3e3f02231eb6a411f6c","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-17 20:25:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9af0d7b44ebeed186a7c80617ec321e118147aca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 20:26:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fb78c38a3c8eef68d95902a3ca98ca57d7b0d81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 20:52:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06ad8b313f354166a32f93bfb039b526\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2dbfa84fd2804e6dae6720cdcba2f2d3 : POST_FAILURE in 3m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8040a5d9626f487e86ea0adb12f7cf76 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"75fdd2a75a1ae0531f1c24e88de35ee185b37a8f","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-17 22:08:45.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"1cb774707726c8484188a6c87e1f12a0ef3d1016","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-24 17:48:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7bb2cb6b8286cc613d3f9ccf22e9b8891485a5f2","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-24 17:48:53.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"d71e95effdeb829e05ed15ff71fde770986ab703","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-24 17:49:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"662062e5b66aec90941d49a8854db72f19a0bc94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 18:28:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465ad2528fc2498b9da92d5a31d30431\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d12d4e492b644722a5350cef14e4cedd : SUCCESS in 4m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bebdf721cd1d48f1bd42b7beeae4bf02 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"02d76131ba2fa4fe484e6cf12d3b5560abdc421c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 18:28:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca490e578c9986b0b47b597b1caeb04fce2f1aee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 18:43:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c727c7496344adf96903a5a9e0c5e00\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bd4c04adbf764ed9862ad9a563aa2e18 : SUCCESS in 5m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4ac04a54bbff4865bebd46975791c16d : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a8d1d872f8cc9b4a5fb362f0fe3a7db716530b85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 18:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0e3a5e68f6d1eed8eba6de68cdcfccd5a7a6a32b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 18:50:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85505c6dd136488bbe29849b3c1ed2ed\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/b5ddbfaf5d7d46fd82bd3314f4d25c2f : SUCCESS in 7m 06s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/db50b91ca837455d8d47586cf3ddb707 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32db852103a5c05de81efe47ae05bad679ee605d","revisions":{"32db852103a5c05de81efe47ae05bad679ee605d":{"kind":"REWORK","_number":1,"created":"2025-06-16 20:13:24.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/12/952712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/952712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/952712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/952712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/952712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/952712/1"}}},"commit":{"parents":[{"commit":"8c9151dbc224bd38d7fc98c69ba74b1ee0004ca3","subject":"Allow branch creation for heterogeneous-distributed-training-framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c9151dbc224bd38d7fc98c69ba74b1ee0004ca3"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-06-16 20:03:59.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-06-16 20:09:21.000000000","tz":-420},"subject":"Switch all Zuul tenants to use niz nodesets","message":"Switch all Zuul tenants to use niz nodesets\n\nThis updates the remaining Zuul tenants (in particular, openstack)\nto use the NIZ nodesets where available.  This will increase our\nuse of NIZ, but some nodepool nodes will still be used.\n\nWe currently have nodesets defined in two places:\n\nopendev/base-jobs: these refer only to labels that are supplied by\nnodepool.  Most Zuul tenants use this.\n\nopendev/zuul-providers: there are two files in this repo where\nnodesets are defined: zuul.d/nodesets.yaml, which contains nodesets\nthat reference labels supplied by zuul-launchers, and\nzuul.d/nodepool-nodesets.yaml, which contains nodesets that reference\nlabels supplied by nodepool.\n\nThis change stops loading nodesets from base-jobs, and instead loads\nnodesets from zuul-providers.  That means that many, but not all,\nthe labels will supplied by zuul-launchers.\n\nIf something goes wrong, we can revert this change to switch all the\ntenant backs to nodepool.  If it goes well, we can continue to\nmigrate nodesets that reference nodepool labels to reference\nzuul-launcher labels instead.\n\nOnce opendev/zuul-providers/nodesets.yaml is empty, we can retire\nthe nodepool servers.\n\nChange-Id: I37579d022578cf35a5d33ab4902b5f0696a32e0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32db852103a5c05de81efe47ae05bad679ee605d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32db852103a5c05de81efe47ae05bad679ee605d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
