)]}'
{"id":"openstack%2Fproject-config~953069","triplet_id":"openstack%2Fproject-config~master~I4955b38c839dee5dad093002ef986174ff451d26","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"38053":{"account":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"last_update":"2025-06-23 12:31:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4955b38c839dee5dad093002ef986174ff451d26","subject":"Allow tag creation for heterogeneous-distributed-training-framework","status":"MERGED","created":"2025-06-23 01:41:09.000000000","updated":"2025-06-23 12:37:03.000000000","submitted":"2025-06-23 12:31:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953069","meta_rev_id":"f10d0dde65b815a2d619ee11fcb46136ea6d4505","_number":953069,"virtual_id_number":953069,"owner":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-23 12:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"all":[{"value":1,"date":"2025-06-23 01:42:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-23 12:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-23 12:25:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-23 02:13:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 12:25:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"50cbae285489b966d16ac0c61e0e12f53ed0a83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-23 01:41:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f573bcdc79f9b103a488f732f3766088463f99b2","author":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"date":"2025-06-23 01:42:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"63f2f86229220e3a6e9aa81fffd8b6c5b36b53b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 02:13:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c49cbc5d864202a0a7f2a8802ff163\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18d36186e95e46af99fdd36a4c08fbd9 : SUCCESS in 4m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fbe605b6a4984a359e3dd090d9c6e80c : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bd443449b68183ec5f8b1dfc5fcf33a30c1cc226","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-23 12:25:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9762e594c9782d295f17caf76572e6062ac15b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 12:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecc06625f95892e875abc2b28ba67617983b9646","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 12:31:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94f2ed4b596b48a89952f5696b7622ec\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/40bbfb8a1c174f73ac192e42c99e967f : SUCCESS in 5m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4e79088a16354c5bb16d5e828d7e6f3d : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9b9d417cfcefc4a2a55dec17fb4f7f2e6ba5c8ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 12:31:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f10d0dde65b815a2d619ee11fcb46136ea6d4505","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 12:37:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7162007cd9ae4a1fbd18965a09fc1bdf\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/86d8d5e714174be196514a159864a142 : SUCCESS in 5m 41s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/293e1e995ee44a3883a100c0cc42a0be : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b533d613cbbb9ad8d287785f52c57eb1a9529562","revisions":{"b533d613cbbb9ad8d287785f52c57eb1a9529562":{"kind":"REWORK","_number":1,"created":"2025-06-23 01:41:09.000000000","uploader":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"},"ref":"refs/changes/69/953069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/953069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/953069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/953069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/953069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/953069/1"}}},"commit":{"parents":[{"commit":"f8e336eb21b2e72063a0f120155e027e0c744b7c","subject":"Merge \"Update neutron grafana dashboard with new FWaaS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f8e336eb21b2e72063a0f120155e027e0c744b7c"}]}],"author":{"name":"tianyutong","email":"tianyutong@sd.chinamobile.com","date":"2025-06-23 01:37:24.000000000","tz":480},"committer":{"name":"tianyutong","email":"tianyutong@sd.chinamobile.com","date":"2025-06-23 01:40:35.000000000","tz":480},"subject":"Allow tag creation for heterogeneous-distributed-training-framework","message":"Allow tag creation for heterogeneous-distributed-training-framework\n\nChange-Id: I4955b38c839dee5dad093002ef986174ff451d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b533d613cbbb9ad8d287785f52c57eb1a9529562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b533d613cbbb9ad8d287785f52c57eb1a9529562"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38053,"name":"tianyutong","display_name":"tianyutong","email":"tianyutong@sd.chinamobile.com","username":"tianyutong"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
