)]}'
{"id":"openstack%2Fproject-config~953819","triplet_id":"openstack%2Fproject-config~master~I86ffbce017cfbe3c07c0efd581c1c8ea8ad9e450","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"9562":{"account":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"last_update":"2025-07-01 13:16:30.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-01 13:02:55.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I86ffbce017cfbe3c07c0efd581c1c8ea8ad9e450","subject":"Replace StarlingX\u0027s CLA enforcement with the DCO","status":"MERGED","created":"2025-06-30 17:39:28.000000000","updated":"2025-07-01 13:27:40.000000000","submitted":"2025-07-01 13:16:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953819","meta_rev_id":"7d32cbe7c231950ffedf6adef8b79dee3b9e0154","_number":953819,"virtual_id_number":953819,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-01 13:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"all":[{"value":1,"date":"2025-06-30 22:45:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-01 13:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-01 13:02:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 17:58:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-30 22:45:03.000000000","updated_by":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"reviewer":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"state":"REVIEWER"}],"messages":[{"id":"1b6f4dcca0b4f4345aadc8f951ca211b32cfc0f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-30 17:39:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d2ae4e540705ba98cbd163f9bbd867c64e1225b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 17:58:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b52c9ff4614dc5a5d4ba2c04034324\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85cdc88703164b5cbec158bf2da42f1f : SUCCESS in 4m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/be2776f2feb448f8b691507bd002a0ce : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ed3760d3dfcbf12a8dd693835cb14b83439567b1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-30 18:01:57.000000000","message":"Patch Set 1:\n\nI\u0027m planning to un-WIP and approve this around 13:00 UTC tomorrow.","accounts_in_message":[],"_revision_number":1},{"id":"79f4e5099f5add8a246b268732ea91ad0a2903d1","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2025-06-30 22:45:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04ef5a00ac90d7d9ccf5873947945016a6163821","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-01 13:02:19.000000000","message":"Set Ready For Review\n\nIt\u0027s time!","accounts_in_message":[],"_revision_number":1},{"id":"2d0b2171a9bf47a3b9bf9dd7494b669b87952b85","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-01 13:02:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acea98138000c7ba0038c99f708a3b216386844b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 13:03:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4581adab5a5dc7922fb6a5889881c6de990dd05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 13:16:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da5394dd3b74120807110fa1a7c8a48\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/355004382edc408d8a66ec353651a09e : SUCCESS in 4m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/26a55a696fe64474847b311ef28e7168 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6277fa009110a60599a3d7d8ea44cfa804303442","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 13:16:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d32cbe7c231950ffedf6adef8b79dee3b9e0154","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 13:27:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345543fbf7bb4e42a57000a5f6c89cd7\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/8c0f0f34c22d4ca7b5ccf808e07fd49f : SUCCESS in 9m 28s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/24467555b5744c9fb13b3068b193c924 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c039fe214dd95cedcf23ef411c9df7439738c636","revisions":{"c039fe214dd95cedcf23ef411c9df7439738c636":{"kind":"REWORK","_number":1,"created":"2025-06-30 17:39:28.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/19/953819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/953819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/953819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/953819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/953819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/953819/1"}}},"commit":{"parents":[{"commit":"86ecd2d6f6f9a66643a017453a5206c21538c4b7","subject":"Reapply \"Switch all Zuul tenants to use niz nodesets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/86ecd2d6f6f9a66643a017453a5206c21538c4b7"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-06-30 17:36:20.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-06-30 17:36:20.000000000","tz":0},"subject":"Replace StarlingX\u0027s CLA enforcement with the DCO","message":"Replace StarlingX\u0027s CLA enforcement with the DCO\n\nThe StarlingX project is switching from the OpenInfra Individual\nContributor License Agreement to the Developer Certificate of\nOrigin, which requires that commit messages include a Signed-Off-By\ntrailer. Any commits generated by automated tooling will need to\ncomply with this requirement as well. Simply including git commit\u0027s\n-s option on the command line should be sufficient to meet this\nrequirement.\n\nChange-Id: I86ffbce017cfbe3c07c0efd581c1c8ea8ad9e450\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c039fe214dd95cedcf23ef411c9df7439738c636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c039fe214dd95cedcf23ef411c9df7439738c636"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
