)]}'
{"id":"openstack%2Fproject-config~953823","triplet_id":"openstack%2Fproject-config~master~Ic869938c17713382a8d83d8d865e9028fd75e104","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-07-01 21:58:48.000000000","reason":"removed on reply"}},"hashtags":["opendev-niz"],"change_id":"Ic869938c17713382a8d83d8d865e9028fd75e104","subject":"Grafana: update zuul status page for NIZ","status":"MERGED","created":"2025-06-30 18:09:06.000000000","updated":"2025-07-01 22:32:07.000000000","submitted":"2025-07-01 22:28:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953823","meta_rev_id":"bcb3b808ae23a264f0a066ac140af3004390ae01","_number":953823,"virtual_id_number":953823,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-01 22:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2025-07-01 21:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2025-06-30 22:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2025-07-01 21:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 18:36:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-30 22:51:57.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"c6fd59b3e884bd932049559cad38b4497a64b968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-30 18:09:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a118650f65b6d4fc5de2a9f409322e8a1163d2b7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-30 18:09:14.000000000","message":"Hashtag added: opendev-niz","accounts_in_message":[],"_revision_number":1},{"id":"7f6a16a4e37208bea4c41f4a5aecc3f048d4e470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 18:36:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627bc40458f5465b947ea57f703474d6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c8baf009dab94f0aba5ace22e877204d : SUCCESS in 18m 27s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/66176232e2f04739a68d1c70d4dbda51 : SUCCESS in 2m 25s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/c067050d9d72432d9acc664e4b8bcfd0 : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1d30f31a9e8c479fc60086498b41b524062a278e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-06-30 22:51:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3522b069fc1bbd14f7d5b4066bea48cabe4ec94","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-01 21:58:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbd2d14a07860638cebec8b93fd498f23f040496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 21:59:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40a32e461f3ad42009bf0baa07cb6bce0d17dc2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 22:28:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c52708b50b4adcad3c49a7c26851b3\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fa3fb2486b5641219bc36efc194af72d : SUCCESS in 28m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fc49b785f76c449aa817088fee0b7ec5 : SUCCESS in 5m 25s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/5b937513ac2b4016ae9c59aa7285d13e : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"831d4efa6c8c30f72f9b9ee4a653d8cf08d91f93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 22:28:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bcb3b808ae23a264f0a066ac140af3004390ae01","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 22:32:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9342c2ed117247f5974a3ec6c0a194fb\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/71e117acee874facb7818b7202357722 : SUCCESS in 2m 51s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/c788517e289e48289ffd719f43a9f4ca : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"074e0c73e8d48ab9a7377ea31eb61ca74e7bcc42","revisions":{"074e0c73e8d48ab9a7377ea31eb61ca74e7bcc42":{"kind":"REWORK","_number":1,"created":"2025-06-30 18:09:06.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/23/953823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/953823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/953823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/953823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/953823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/953823/1"}}},"commit":{"parents":[{"commit":"86ecd2d6f6f9a66643a017453a5206c21538c4b7","subject":"Reapply \"Switch all Zuul tenants to use niz nodesets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/86ecd2d6f6f9a66643a017453a5206c21538c4b7"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-06-30 18:08:20.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-06-30 18:08:52.000000000","tz":-420},"subject":"Grafana: update zuul status page for NIZ","message":"Grafana: update zuul status page for NIZ\n\nThis switches the node requests and node state graphs to use\nnodepool-in-zuul metrics instead of nodepool.\n\nDepends-On: https://review.opendev.org/953805\nChange-Id: Ic869938c17713382a8d83d8d865e9028fd75e104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/074e0c73e8d48ab9a7377ea31eb61ca74e7bcc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/074e0c73e8d48ab9a7377ea31eb61ca74e7bcc42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
