)]}'
{"id":"openstack%2Fproject-config~953825","triplet_id":"openstack%2Fproject-config~master~I69b6193db58a61aa8f09af8a00c5d5403ddf929f","project":"openstack/project-config","branch":"master","topic":"new-zks","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-06-30 22:23:12.000000000","reason":"removed on reply"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-06-30 22:54:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69b6193db58a61aa8f09af8a00c5d5403ddf929f","subject":"Limit grafana graphs for zookeeper to zk01-03","status":"MERGED","created":"2025-06-30 18:58:25.000000000","updated":"2025-06-30 22:57:32.000000000","submitted":"2025-06-30 22:54:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953825-new-zks","meta_rev_id":"732b228dd9cc6c2f3e8a63903db85ba8befeb1e6","_number":953825,"virtual_id_number":953825,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 22:54:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2025-06-30 22:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2025-06-30 22:23:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 19:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-30 22:17:00.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"b9c06bd61c08825be9fa3ff53adfae0513b2d773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-06-30 18:58:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9a1691a20371e8b6cc5882b2e8cc82f697fa837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 19:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce9cda6107746fc9cd99868bf3bd822\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cfc332ad38c148ef8c38da5f91cce382 : SUCCESS in 17m 21s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e3e8b119ef5b48bc8dec3ea8413d52d5 : SUCCESS in 4m 59s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/79be1934109646f186129c6fdba84c88 : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b0471d41c8d8add1fd813a32e1ef25cc212ab2b6","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-30 22:17:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1144e39eb59b588c9b410e293cd1001e56e6a490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-06-30 22:18:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"77c537dd88fef221b517ddc712a9601946dd7921","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-06-30 22:23:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d75e2e62a4ff5a02918497707ffc90f56c8cb960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:38:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b188e2bd648c4aef9e829cb3199ab02b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c7582d5a1fb449ed9f4b1f2310b89d57 : SUCCESS in 14m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/837a95f461104796b1f8909745ba8e24 : SUCCESS in 4m 29s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/e7f70913b8b84f5394eaed9aa16995c6 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"eb821544d8f9e8f7df2406198cc9d6d1d8c9793e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:39:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bdddba5481bb67577107166ecf7f44b017b922d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:54:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f1c24431d846c88962d2329b5721e6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/298ccef966f54fdb8ad0512021c60374 : SUCCESS in 14m 33s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b24b845661b0448e97b9e31049ca24e1 : SUCCESS in 2m 20s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/8137f17281f145d98f78f9b5635335a7 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":2},{"id":"220496ad2e0f0afd2fb8d3c903cf880475706332","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:54:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"732b228dd9cc6c2f3e8a63903db85ba8befeb1e6","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:57:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51e054e2c00455497236bd34f24b975\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/27227f74aa0d487d9bcc45599dc1f1b3 : SUCCESS in 3m 03s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/26b09c07426b4052be560b2ade934a25 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3a348f97ddce055d4b764b7613505eeb93e8080","revisions":{"d7091520b2ec6bde4eb3e1e990d51cb416c02718":{"kind":"REWORK","_number":1,"created":"2025-06-30 18:58:25.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/25/953825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/953825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/953825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/953825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/953825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/953825/1"}}},"commit":{"parents":[{"commit":"86ecd2d6f6f9a66643a017453a5206c21538c4b7","subject":"Reapply \"Switch all Zuul tenants to use niz nodesets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/86ecd2d6f6f9a66643a017453a5206c21538c4b7"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-06-30 18:57:45.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-06-30 18:57:45.000000000","tz":-420},"subject":"Limit grafana graphs for zookeeper to zk01-03","message":"Limit grafana graphs for zookeeper to zk01-03\n\nWe\u0027re replacing zk04-06 (which are the current servers in the graphs)\nwith zk01-03. Update the limits in the graphs to match zk01 to zk03 now\nthat those servers are in place.\n\nChange-Id: I69b6193db58a61aa8f09af8a00c5d5403ddf929f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7091520b2ec6bde4eb3e1e990d51cb416c02718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7091520b2ec6bde4eb3e1e990d51cb416c02718"}]},"branch":"refs/heads/master"},"e3a348f97ddce055d4b764b7613505eeb93e8080":{"kind":"REWORK","_number":2,"created":"2025-06-30 22:18:57.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/25/953825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/953825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/953825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/953825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/953825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/953825/2"}}},"commit":{"parents":[{"commit":"86ecd2d6f6f9a66643a017453a5206c21538c4b7","subject":"Reapply \"Switch all Zuul tenants to use niz nodesets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/86ecd2d6f6f9a66643a017453a5206c21538c4b7"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-06-30 18:57:45.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-06-30 22:18:51.000000000","tz":-420},"subject":"Limit grafana graphs for zookeeper to zk01-03","message":"Limit grafana graphs for zookeeper to zk01-03\n\nWe\u0027re replacing zk04-06 (which are the current servers in the graphs)\nwith zk01-03. Update the limits in the graphs to match zk01 to zk03 now\nthat those servers are in place.\n\nChange-Id: I69b6193db58a61aa8f09af8a00c5d5403ddf929f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3a348f97ddce055d4b764b7613505eeb93e8080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3a348f97ddce055d4b764b7613505eeb93e8080"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
