)]}'
{"id":"openstack%2Fproject-config~953973","triplet_id":"openstack%2Fproject-config~master~I0d3f28d46fc4bf272bcd3246ce18476073220d73","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-07-02 17:37:25.000000000","reason":"Change was submitted"}},"hashtags":["opendev-niz"],"change_id":"I0d3f28d46fc4bf272bcd3246ce18476073220d73","subject":"Revise zuul-launcher dashboards","status":"MERGED","created":"2025-07-02 16:18:08.000000000","updated":"2025-07-02 23:11:06.000000000","submitted":"2025-07-02 17:37:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953973","meta_rev_id":"bcd0475a6c0ea87dbf4ccb62d03318cfb204535c","_number":953973,"virtual_id_number":953973,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 17:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-07-02 17:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-02 17:12:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-02 17:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-02 16:56:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 17:04:46.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-07-02 17:12:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d3c0be5bcdf53983c843e6effe12fe6c1676e6a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-02 16:18:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"321f244fa954c36b4a859d4e8b9a847b4f422e5c","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-02 16:23:10.000000000","message":"Hashtag added: opendev-niz","accounts_in_message":[],"_revision_number":1},{"id":"5eef30d3309a80c5980c6a5b0f4a7ab079d72839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 16:56:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01880adec8bf4b4f9ab404b837513118\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/93d3d402ff5d4055843427bff7a760f2 : SUCCESS in 22m 06s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea9d176144434fa19ad09368cb996fa3 : SUCCESS in 2m 59s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/c59928486bb348ef80841e3e1d704909 : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fcf456c080c071cfc9fd523bef8032d3d6025dbf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-02 17:04:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cce435490bf4ef027c880b3de3450e79163a3dd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-02 17:12:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e65ec2592b179f52e2b337ab2af2304597ca12b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 17:13:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37aa7226855ec3ba84e7cd4535cbf5206aba2716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 17:37:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c1cb6ba5efe40779a33cde23ef6291e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/889a5e69bb014b67935bada1495d7c8e : SUCCESS in 22m 37s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ee022ac59d8e4400b979d4b4d725a7db : SUCCESS in 5m 35s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/3b99c481018b4d658c717d2a4f7f82c7 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e85e453401f3d40c3d392f23d61762f855bad523","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 17:37:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e9e6e2fa62f7a772195d215b65bed1b71c77a71","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 17:41:01.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e7978c0efcd4ae98f33af10c9359bb0\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/1cabcd4cb0024b33899dab6f296d3301 : SUCCESS in 3m 17s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/ebd85e269d1242bfa96fcf8fe504ed99 : FAILURE in 1m 27s","accounts_in_message":[],"_revision_number":1},{"id":"de6972e1c0fc850233d61ed41ca47675e646a707","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 22:36:04.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5bcc68d9bb2488db40f43f14e4b27e5\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/c1987063233046eea7be641c8da16cb8 : SUCCESS in 3m 12s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/b04bb3818e374095b4ad267785f11750 : FAILURE in 1m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bcd0475a6c0ea87dbf4ccb62d03318cfb204535c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 23:11:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76a41b0d0ac84df39a892feed5f33874\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/c7628f9118f246129137e3148af25bd2 : SUCCESS in 2m 59s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/e717abf97b6b44df8fb420c0133eae53 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d06a48229fb52bbf6bfc3ce1e0797203b332aa9c","revisions":{"d06a48229fb52bbf6bfc3ce1e0797203b332aa9c":{"kind":"REWORK","_number":1,"created":"2025-07-02 16:18:08.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/73/953973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/953973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/953973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/953973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/953973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/953973/1"}}},"commit":{"parents":[{"commit":"f6cd3014b2ffc7e1b207144d42641d7afabebccc","subject":"Fix zuul status node requests graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6cd3014b2ffc7e1b207144d42641d7afabebccc"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-02 16:14:06.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-02 16:14:06.000000000","tz":-420},"subject":"Revise zuul-launcher dashboards","message":"Revise zuul-launcher dashboards\n\nThis makes the following changes to the zuul-launcher dashboards:\n* Adds gauges that show current/max quota values for\n  instances, cores, ram\n* Converts the template from YAML to JSON due to the complexity\n  required for the gauges\n* Since it\u0027s not easy to template which quota values are present,\n  two templates are provided: one with all 3, and another with only\n  instances and ram (for rax)\n* Updates the test node history graph to stack (so that we can compare\n  the total of building+ready+used+etc to the max line (the max line\n  is not stacked)\n* Adds a 10% opaque background to all lines except max to make\n  the stacking more apparent\n* Adds a \"requested\" node state (which is new in niz)\n* Removes the unused \"deleted\" singlestat and replaces it with\n  \"requested\"\n\nChange-Id: I0d3f28d46fc4bf272bcd3246ce18476073220d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d06a48229fb52bbf6bfc3ce1e0797203b332aa9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d06a48229fb52bbf6bfc3ce1e0797203b332aa9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
