)]}'
{"id":"openstack%2Fproject-config~954035","triplet_id":"openstack%2Fproject-config~master~I70ecb4c4f17187efb0e3b79d488c3e397b231769","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-03 13:37:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I70ecb4c4f17187efb0e3b79d488c3e397b231769","subject":"Drop lingering ubuntu-xenial jobs","status":"MERGED","created":"2025-07-03 13:20:10.000000000","updated":"2025-07-03 13:40:13.000000000","submitted":"2025-07-03 13:37:49.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954035","meta_rev_id":"8e7099cfbc1a7998b48644306534bc1a08941a0d","_number":954035,"virtual_id_number":954035,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2025-07-03 13:37:49.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":0,"date":"2025-07-03 13:27:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2025-07-03 13:37:49.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2025-07-03 13:37:49.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-03 13:20:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-03 13:36:15.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2025-07-03 13:37:49.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"},{"updated":"2025-07-03 13:40:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b016c79e6e64f99dcccfe9b02c36090d0153962a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-03 13:20:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de489f18a1e9314385005cb0cb8725eaf23d908f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:20:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The nodeset \"ubuntu-xenial\" was not found.\n\nThe problem appears in the the \"publish-openstack-sphinx-docs-base\" job stanza:\n\n  job:\n      name: publish-openstack-sphinx-docs-base\n      nodeset: ubuntu-xenial\n      description: |\n        Base job for publishing sphinx documentation to\n        /afs/.openstack.org/docs/.\n  \n        This job is obsolete, use ``publish-openstack-tox-docs-base`` instead.\n      abstract: true\n  ...\n\n  in \"openstack/project-config/zuul.d/jobs.yaml@master\", line 81","accounts_in_message":[],"_revision_number":1},{"id":"10ba91a669c097fac2f48db6348c4a46311d1711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-03 13:26:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ea5e526697d2476a225ad918ec2bb2252472e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:27:05.000000000","message":"Patch Set 2: Verified-1\n\nUnable to freeze job graph: The nodeset \"ubuntu-xenial\" was not found.","accounts_in_message":[],"_revision_number":2},{"id":"c862c0d362e02ffae69c78758f696b83e56e1ee0","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2025-07-03 13:36:15.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_22348\u003e with the following votes:\n\n* Verified-1 by \u003cGERRIT_ACCOUNT_22348\u003e\n","accounts_in_message":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"4ed6448e089327b5bd9de730750319e904f4c5eb","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2025-07-03 13:37:49.000000000","message":"Patch Set 2: Code-Review+2 Verified+2 Workflow+1\n\nBypassing Zuul to merge this since it will not be speculatively tested and is blocked by the error it addresses.","accounts_in_message":[],"_revision_number":2},{"id":"a64e049c5cf9ceec35b35bd51ef516699c85f1e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2025-07-03 13:37:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8e7099cfbc1a7998b48644306534bc1a08941a0d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:40:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbabf1cd9ed0442780b4fa38ce35b408\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/67e643a748004b9fb8ed286c61d2f342 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b976490d65ff1abfa9450ebfc993386fb36dc645","revisions":{"fbf899fee7a5f2fc99d18f7a9546f0aeef75dc95":{"kind":"REWORK","_number":1,"created":"2025-07-03 13:20:10.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/35/954035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/954035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/954035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/954035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/954035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/954035/1"}}},"commit":{"parents":[{"commit":"d06a48229fb52bbf6bfc3ce1e0797203b332aa9c","subject":"Revise zuul-launcher dashboards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d06a48229fb52bbf6bfc3ce1e0797203b332aa9c"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-03 13:18:07.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-03 13:18:07.000000000","tz":0},"subject":"Drop lingering ubuntu-xenial bindep job","message":"Drop lingering ubuntu-xenial bindep job\n\nWe stopped running this in bindep, but forgot that project-config\nwas still using it to check changes to the fallback packages list.\n\nChange-Id: I70ecb4c4f17187efb0e3b79d488c3e397b231769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fbf899fee7a5f2fc99d18f7a9546f0aeef75dc95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fbf899fee7a5f2fc99d18f7a9546f0aeef75dc95"}]},"branch":"refs/heads/master"},"b976490d65ff1abfa9450ebfc993386fb36dc645":{"kind":"REWORK","_number":2,"created":"2025-07-03 13:26:18.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/35/954035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/954035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/954035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/954035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/954035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/954035/2"}}},"commit":{"parents":[{"commit":"d06a48229fb52bbf6bfc3ce1e0797203b332aa9c","subject":"Revise zuul-launcher dashboards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d06a48229fb52bbf6bfc3ce1e0797203b332aa9c"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-03 13:18:07.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-03 13:24:03.000000000","tz":0},"subject":"Drop lingering ubuntu-xenial jobs","message":"Drop lingering ubuntu-xenial jobs\n\nWe stopped testing on ubuntu-xenial in bindep, but forgot that\nproject-config was still using it to check changes to the fallback\npackages list.\n\nAlso publish-openstack-sphinx-docs and its parent have been\ndeprecated for a long time and seem to no longer be in use, but rely\non an ubuntu-xenial nodeset which we have ceased providing.\n\nChange-Id: I70ecb4c4f17187efb0e3b79d488c3e397b231769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b976490d65ff1abfa9450ebfc993386fb36dc645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b976490d65ff1abfa9450ebfc993386fb36dc645"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
