)]}'
{"id":"openstack%2Fproject-config~954246","triplet_id":"openstack%2Fproject-config~master~I0f23bb80472e236f53124334290943394e148efd","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-07-09 20:56:42.000000000","reason":"Change was submitted"}},"hashtags":["opendev-niz"],"change_id":"I0f23bb80472e236f53124334290943394e148efd","subject":"Update Zuul status node graphs","status":"MERGED","created":"2025-07-07 13:43:57.000000000","updated":"2025-07-09 21:00:11.000000000","submitted":"2025-07-09 20:56:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954246","meta_rev_id":"cdbfbe215fdd7d82862f84c3de319e2451c42e18","_number":954246,"virtual_id_number":954246,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-09 20:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-07-09 20:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-09 20:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-07-09 20:36:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 14:02:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-09 20:14:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-07-09 20:16:53.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-09 20:36:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"6d153513e18072c63cb6a26693b1adb57c2f6b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-07 13:43:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41c0ab927e06407b5bba629859bd9515bbdf41cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:02:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd0217e7cf444f4a896c2ae99a81f79\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8a313ed04e794d65a58d5124fbfd175c : SUCCESS in 16m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dba3c73f6ac74261b74266d93cda065e : SUCCESS in 4m 58s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/57bea3be98d949bb93bd8041989892d3 : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":1},{"id":"85043849ca0be043cc9f1023f9153ce8369fad6b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-09 20:09:28.000000000","message":"Hashtag added: opendev-niz","accounts_in_message":[],"_revision_number":1},{"id":"b101d2723e6369d478f920e4990a95a944b26f31","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-09 20:14:38.000000000","message":"Patch Set 1:\n\nThe commit message doesn\u0027t mention the deleting state removal, was this intentional?","accounts_in_message":[],"_revision_number":1},{"id":"1e2f71a814adb4d494d321c18d8d88b13b2e4562","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-09 20:15:47.000000000","message":"Patch Set 1:\n\nYes, there is no deleting state.","accounts_in_message":[],"_revision_number":1},{"id":"9595c2df9a3d158b9e32c5271299b6b14780753d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-09 20:16:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30f27907b344fd3286507225a481b9ea704a3d28","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-09 20:36:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8b70870035166f926d382aa0d8ac06be71f4378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 20:36:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f8023f20f80eb313dabf2a997a8f9974d97753b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 20:56:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b9c42a1269d410c8aeb77e3f97e4069\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6328c3521f65430b815005189e5ef239 : SUCCESS in 18m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/103cd311fe934ddf953d3df19c0d07d2 : SUCCESS in 4m 52s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/347fff6c1dc143eaa4103a98addd5def : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8ae67e2be2658b5c4c910a16c190c5755748b9d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 20:56:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cdbfbe215fdd7d82862f84c3de319e2451c42e18","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 21:00:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce81e6c5520d488a9706973e36f2e6fd\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/0e99195e15ce4bd4b369b2162b83719b : SUCCESS in 3m 04s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/4c9b6f884b644134a7bded09b6b15463 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a75b92eb92d0994ad9e99e3530bd876e256dfe2f","revisions":{"a75b92eb92d0994ad9e99e3530bd876e256dfe2f":{"kind":"REWORK","_number":1,"created":"2025-07-07 13:43:57.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/46/954246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/46/954246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/954246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/954246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/954246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/46/954246/1"}}},"commit":{"parents":[{"commit":"53d254bc73e95a72cdc5cec89d34c8c3939639ed","subject":"Remove superfluous quoting on commit trailer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53d254bc73e95a72cdc5cec89d34c8c3939639ed"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-07 13:41:55.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-07 13:43:38.000000000","tz":-420},"subject":"Update Zuul status node graphs","message":"Update Zuul status node graphs\n\nThis updates the \"Test Nodes\" and \"Node Requessts\" graphs on the\nZuul status dashboard to be more consistent:\n\n* The node requests graph is now stacked, like the test nodes graph,\n  so that the topmost line is the total.\n* The popover for the node requests graph will now show all the\n  values.\n* The test nodes graph now also shows requested nodes (a new state).\n* The y-axis attributes for the node requests graph do not seem to\n  be necessary, so are removed.\n\nChange-Id: I0f23bb80472e236f53124334290943394e148efd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a75b92eb92d0994ad9e99e3530bd876e256dfe2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a75b92eb92d0994ad9e99e3530bd876e256dfe2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
