)]}'
{"id":"openstack%2Fproject-config~954374","triplet_id":"openstack%2Fproject-config~master~Ia22b30b071c12c43c75ffb7a90923556ad0d2c7e","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ia22b30b071c12c43c75ffb7a90923556ad0d2c7e","subject":"Remove CLA enforcement from all projects and lock","status":"MERGED","created":"2025-07-08 21:04:04.000000000","updated":"2025-07-18 14:56:29.000000000","submitted":"2025-07-18 14:20:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954374","meta_rev_id":"ad61e62627fbe47d836dc10e9d9cabde883a76ea","_number":954374,"virtual_id_number":954374,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-18 14:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-18 14:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-18 14:05:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-08 21:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc8a868ed7e64d8742ee3be94242f4c01919bd79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-08 21:04:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5971dc24a83897d42e6090502270eb08ca06cce9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-08 21:05:17.000000000","message":"Set Work In Progress\n\nWe should announce this change, in order to give affected projects some advance notice in case they want to make sure they enact other measures, such as adding RequireSignedOffBy or something.","accounts_in_message":[],"_revision_number":1},{"id":"21760aafa87d81bcc010985e7483d248f51b58c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 21:09:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfac32d9b642450d935746332d60bab4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/193f4a1deb45484292a1b27504a67335 : SUCCESS in 3m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1fff01e8b98845e5abbc74f7106e9d59 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"22e65e6e74faa5ae63d58daf805b0d5398bdb596","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-15 18:05:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4c352f4373b7f02b3a7741dd052fa736c15ec424","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-15 18:07:25.000000000","message":"Patch Set 2:\n\nrebased after merging I4bd97b4356924f89b91783cefa296116957aa944 which conflicted","accounts_in_message":[],"_revision_number":2},{"id":"eb84fb6476cffec4192bd8856f957091eb277b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 18:09:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef1d741960f4b51a258eb0de01c4fc2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d12d1df5c6b74ad5b4ab859d3c7bae51 : SUCCESS in 2m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8cf8840fa18747ccb917a66d34aedf96 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c246f8395fd82b23a4dd823cb737ee0f33c2bac5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-18 14:04:17.000000000","message":"Set Ready For Review\n\nToday\u0027s the day!","accounts_in_message":[],"_revision_number":2},{"id":"9fdb4636e2b97a1d66febb2f714bd430b7f02188","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-18 14:05:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI\u0027m self-approving this, it was announced over a week in advance on the service-announce mailing list, and discussed at the weekly meeting too.","accounts_in_message":[],"_revision_number":2},{"id":"fc69580b69e75eefed9d4bad04efe0b85a9b9fd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 14:05:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e90b249494e8d6c6f35e0581276188ae37fbd99b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 14:20:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dc9d39d6032427aa06ac95d9d592f34\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/214d157cec914c09a8938a849d4ab847 : SUCCESS in 4m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d314f42786534a16b250d6e8f297790a : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f596e9aa3975f0b8c9ce3221357d3c1b6cefc311","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 14:20:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ad61e62627fbe47d836dc10e9d9cabde883a76ea","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 14:56:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e747ba1c4ba40dfa882ede726017c28\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/907b2c0d2c4546849febc4af125346ef : SUCCESS in 35m 46s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0173b889b2cb4c8699b6bcec6150e02e : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f497a0599ae98baa73e77f7f2d704cb7f77e934","revisions":{"4a847b9e378775db12b13481ccb9b0def5305e5f":{"kind":"REWORK","_number":1,"created":"2025-07-08 21:04:04.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/74/954374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/954374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/954374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/954374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/954374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/954374/1"}}},"commit":{"parents":[{"commit":"53d254bc73e95a72cdc5cec89d34c8c3939639ed","subject":"Remove superfluous quoting on commit trailer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53d254bc73e95a72cdc5cec89d34c8c3939639ed"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-08 20:59:00.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-08 20:59:00.000000000","tz":0},"subject":"Remove CLA enforcement from all projects and lock","message":"Remove CLA enforcement from all projects and lock\n\nThe configured contributor license agreements in our Gerrit are\nreferent to the Open Infrastructure Foundation incorporated in\nDelaware, which is in the process of being closed down after its\noperations joined the Linux Foundation. As such, they are no longer\nvalid for future contributions (and it\u0027s questionable whether they\nwere ever valid for non-OpenInfra projects anyway). OpenInfra\nprojects which were previously relying on these have since switched\nto the DCO with a different enforcement mechanism.\n\nRemove CLA enforcement from all remaining repositories and lock down\nour ACL linter to reject future additions, so that we can\nunconfigure support in our Gerrit.\n\nChange-Id: Ia22b30b071c12c43c75ffb7a90923556ad0d2c7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a847b9e378775db12b13481ccb9b0def5305e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a847b9e378775db12b13481ccb9b0def5305e5f"}]},"branch":"refs/heads/master"},"4f497a0599ae98baa73e77f7f2d704cb7f77e934":{"kind":"REWORK","_number":2,"created":"2025-07-15 18:05:41.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/74/954374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/954374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/954374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/954374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/954374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/954374/2"}}},"commit":{"parents":[{"commit":"99b55ea1862ec89e29d047b72530227838a26f05","subject":"Merge \"DCO enforcement for all OpenInfra Foundation repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/99b55ea1862ec89e29d047b72530227838a26f05"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-08 20:59:00.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-15 18:04:44.000000000","tz":0},"subject":"Remove CLA enforcement from all projects and lock","message":"Remove CLA enforcement from all projects and lock\n\nThe configured contributor license agreements in our Gerrit are\nreferent to the Open Infrastructure Foundation incorporated in\nDelaware, which is in the process of being closed down after its\noperations joined the Linux Foundation. As such, they are no longer\nvalid for future contributions (and it\u0027s questionable whether they\nwere ever valid for non-OpenInfra projects anyway). OpenInfra\nprojects which were previously relying on these have since switched\nto the DCO with a different enforcement mechanism.\n\nRemove CLA enforcement from all remaining repositories and lock down\nour ACL linter to reject future additions, so that we can\nunconfigure support in our Gerrit.\n\nChange-Id: Ia22b30b071c12c43c75ffb7a90923556ad0d2c7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f497a0599ae98baa73e77f7f2d704cb7f77e934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f497a0599ae98baa73e77f7f2d704cb7f77e934"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
