)]}'
{"id":"openstack%2Fproject-config~954672","triplet_id":"openstack%2Fproject-config~master~I4bd97b4356924f89b91783cefa296116957aa944","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-15 17:47:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I4bd97b4356924f89b91783cefa296116957aa944","subject":"DCO enforcement for all OpenInfra Foundation repos","status":"MERGED","created":"2025-07-10 22:14:44.000000000","updated":"2025-07-15 18:04:44.000000000","submitted":"2025-07-15 17:57:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954672","meta_rev_id":"227160796a5254c0c5f80f19dda1e223e3819040","_number":954672,"virtual_id_number":954672,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-15 17:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-07-10 22:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2025-07-15 17:48:43.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-15 17:47:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 22:17:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-07-10 22:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46232b34927c6b717d8b2c09c08b2fd2a42e00eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-10 22:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bcb2a2b609dd04bae10fbc5b4656ea5c570eb07","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-10 22:17:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e459a5a2a144c4b4dbf7c2e7991af150c5c1b505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 22:23:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dbf5631990543b790fb08c7cb05169e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f8502e28c6044580abb695bf61a7305a : SUCCESS in 4m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/48e2a086924f442ba6c7bf8413e022b6 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"aec55c6f3f43acc4446f11cf0833214185de5989","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-15 17:47:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bbf89e96a17d330470a59a3a0b7dcc2e29b2a15d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 17:48:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a65432eb4a91c8f38eb1e229afdbe9d5ea7fbf9d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-15 17:48:43.000000000","message":"Patch Set 1: -Code-Review\n\nSelf-approving now since it\u0027s going to conflict with the removal we\u0027re preparing to merge at the end of the week and I want to make sure to take the time to get it correctly updated.","accounts_in_message":[],"_revision_number":1},{"id":"dcb904f09636f7981e9145739e7805945574c835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 17:57:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74bcf036318d45998c0b72e9a6b24b53\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/95776455240c422782ab99e4e64a74b4 : SUCCESS in 4m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b287fc1694df40ffad286ded1c5dd1a4 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"50c9a0de3f8ac638999bd47c95ed919d67f784a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 17:57:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"227160796a5254c0c5f80f19dda1e223e3819040","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 18:04:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baebf9e85d514c50a926dd8d6568c46c\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/f47dce752542435cbed260fdc70bb5fe : SUCCESS in 6m 29s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/3d04caa8f1314854af8dd29d9a1c3baf : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd41fa1b49704d98ed62cb7bb0cbf933576dd682","revisions":{"fd41fa1b49704d98ed62cb7bb0cbf933576dd682":{"kind":"REWORK","_number":1,"created":"2025-07-10 22:14:44.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/72/954672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/954672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/954672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/954672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/954672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/954672/1"}}},"commit":{"parents":[{"commit":"3434225032f45979232b611a60a0602e7a8e4426","subject":"Merge \"Update Zuul status node graphs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3434225032f45979232b611a60a0602e7a8e4426"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-10 22:11:51.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-07-10 22:11:51.000000000","tz":0},"subject":"DCO enforcement for all OpenInfra Foundation repos","message":"DCO enforcement for all OpenInfra Foundation repos\n\nOpendev hosts a number of repositories for OpenInfra Foundation\nefforts which are not themselves Open Infrastructure Projects. These\nrepositories were enforcing the old foundation CLA which is no\nlonger valid. After consulting the foundation\u0027s executive\nmanagement, it was decided they should use the DCO instead like\nAirship/Kata/OpenStack/StarlingX are already doing.\n\nChange-Id: I4bd97b4356924f89b91783cefa296116957aa944\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd41fa1b49704d98ed62cb7bb0cbf933576dd682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd41fa1b49704d98ed62cb7bb0cbf933576dd682"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
