)]}'
{"id":"openstack%2Fproject-config~954758","triplet_id":"openstack%2Fproject-config~master~I7a557026728e5ac458ae8780043992a54ae6d87e","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-07-11 16:25:23.000000000","reason":"Change was submitted"}},"hashtags":["opendev-niz"],"change_id":"I7a557026728e5ac458ae8780043992a54ae6d87e","subject":"Remove bionic/focal arm64 wheel jobs","status":"MERGED","created":"2025-07-11 16:09:53.000000000","updated":"2025-07-11 16:27:51.000000000","submitted":"2025-07-11 16:25:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954758","meta_rev_id":"d8950925d693be8fb7570bd5cda34b47eaaf47e4","_number":954758,"virtual_id_number":954758,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 16:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2025-07-11 16:14:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-11 16:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-11 16:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 16:12:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-11 16:14:26.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2025-07-11 16:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"432188bf733c6e17668bf9ec3cc0898ef3b067ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-11 16:09:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf330202a68dd943f1a54e834d265321a244d9c6","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-11 16:10:34.000000000","message":"Hashtag added: opendev-niz","accounts_in_message":[],"_revision_number":1},{"id":"d1beed08ae8e95ca9c18bbf4ed48bef1621339b6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-11 16:12:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8817c3dfde95cccc5c4cd4c4eda2b4db7578da24","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2025-07-11 16:14:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"63f70ecd9eae97b7f07f32e73ca317440eff72f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:16:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa62147c580643d8a82fdd04eb49a0e3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/acda59e7d8544f398dcee842a583ca6e : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"17730177f61fb296588c0d15b590089e1086d9ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:17:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bba2df74b12ae52b247a2c701c50025c0a3ef711","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:25:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52705dbe24a246c393d6f625f4edac5f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e0125d21b72b4c5c9f86b7b25d199d63 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d24ae20a40513f535e86e8e470ba1459161c6610","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:25:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d8950925d693be8fb7570bd5cda34b47eaaf47e4","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:27:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3150e7d73284feb8abda6dbefa7d84b\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/eaebef2f1bdb49e384c1ce91f2f2244b : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecce303919687b3bdc5a52ddd850bc6c76f2c8c1","revisions":{"ecce303919687b3bdc5a52ddd850bc6c76f2c8c1":{"kind":"REWORK","_number":1,"created":"2025-07-11 16:09:53.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/58/954758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/58/954758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/954758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/954758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/58/954758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/58/954758/1"}}},"commit":{"parents":[{"commit":"a75b92eb92d0994ad9e99e3530bd876e256dfe2f","subject":"Update Zuul status node graphs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a75b92eb92d0994ad9e99e3530bd876e256dfe2f"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-11 16:09:00.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-11 16:09:00.000000000","tz":-420},"subject":"Remove bionic/focal arm64 wheel jobs","message":"Remove bionic/focal arm64 wheel jobs\n\nThere are no consumers of this architecture now (except these jobs)\nso remove them so we can remove the nodepool config.\n\nChange-Id: I7a557026728e5ac458ae8780043992a54ae6d87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecce303919687b3bdc5a52ddd850bc6c76f2c8c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecce303919687b3bdc5a52ddd850bc6c76f2c8c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
