)]}'
{"id":"openstack%2Fproject-config~956184","triplet_id":"openstack%2Fproject-config~master~I149a0d146920424f6f4d202e5d915406b92e9bb2","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-08-20 04:22:33.000000000","reason":"Change was submitted"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-08-06 16:44:05.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}},"hashtags":["opendev-niz"],"change_id":"I149a0d146920424f6f4d202e5d915406b92e9bb2","subject":"Remove nodepool configuration/elements","status":"MERGED","created":"2025-07-30 18:02:54.000000000","updated":"2025-08-20 04:34:33.000000000","submitted":"2025-08-20 04:22:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956184","meta_rev_id":"88a464d85875e1febf671b06961406b761767ee6","_number":956184,"virtual_id_number":956184,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 04:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-08-06 19:38:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-08-20 04:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2025-08-20 04:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 18:05:20.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-07-30 18:08:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-20 04:14:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b6981660039855a48d63317f9f8c1dda768d042f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-30 18:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfb5e3c6039dc4723540efc3d928617a56391aa5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-30 18:05:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa71f821b4ff5a437fff00d576425a6f3037333b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-07-30 18:07:12.000000000","message":"Hashtag added: opendev-niz","accounts_in_message":[],"_revision_number":1},{"id":"719fb01411655acbe67068df41dd0fd0aec11d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 18:08:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bcb36a8a08a433ea245bf04d76f21eb\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ac1bee0ee7fb41538c014eab4d0fc764 : SUCCESS in 4m 06s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/d4e10c777bec406f8f31551f2373623f : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ee8fddb1b12c8352d3ae1c7b99298c6313883dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-08-06 16:42:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ad66d1832b26a076ecb0c333d1ff70e07257b9f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-06 16:44:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bec70ac4108f1a98d4f2c7406bab257590fb38bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:46:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd13df1fa3c44b689e5e2d84507e889c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7223eab2dec04d42912d54b5c694ccd1 : SUCCESS in 2m 22s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/e9b81ecba26e41eb9776d386c4a3fc19 : FAILURE in 1m 47s","accounts_in_message":[],"_revision_number":2},{"id":"7c22893fe77bdf9898761276192e7687dcc46e06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-08-06 19:38:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8d191b9e678e6ab433382edc52a54456ce918f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:53:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/213038c4ae3f4a62a2a25dc7586745ae\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e8aa19de46de4287a0a104b687b1599e : SUCCESS in 4m 12s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/85ce7800a871443cb68df06e2a3686c4 : FAILURE in 1m 57s","accounts_in_message":[],"_revision_number":3},{"id":"20ed94a8874d1d2835cd46e289ff2ecf16681594","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-08-06 20:36:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b98b85b7f1c6c26fb8d05f0ba4ca32a43e561544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:45:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aac4fdfaf844cf7ba43279c147a5b7c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/740a0573e017498b83ce4bab80010e75 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"9195c9a603ca14e3c0d825fc500993644a86a4af","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-20 04:14:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ff3224ed696398f100eb326bd2d06136130d8b94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 04:14:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6169338e9c1a89acf33af2483d171a5fbca38443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 04:22:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018f4bf4288840b0887fc9ed27be12a5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b283d439562442ee8b69f98ef696c5f1 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8cdc7cb6bb4faedbb649426e2df3092228778d00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 04:22:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"88a464d85875e1febf671b06961406b761767ee6","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 04:34:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb51195629ae494686e91f3bc5d300fb\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/95b39ce073d6481bb7705e4e1dd989c9 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b22bdcf480bf91562709da61252f14ba3c2f6f46","revisions":{"b7ed9a22e97183022886c7c5b0b2b7c61ddf27fb":{"kind":"REWORK","_number":1,"created":"2025-07-30 18:02:54.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/84/956184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/956184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/956184/1"}}},"commit":{"parents":[{"commit":"9d2c590bb986ebcf29ef74593a7cf32f55bd82c3","subject":"Merge \"Remove CLA enforcement from all projects and lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d2c590bb986ebcf29ef74593a7cf32f55bd82c3"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-30 18:01:19.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-30 18:01:19.000000000","tz":-420},"subject":"Remove nodepool configuration/elements","message":"Remove nodepool configuration/elements\n\nThis removes the nodepool configuration and DIB elements.  Our use\nof nodepool has been replaced by zuul-launcher, and our elements have\nmoved to the zuul-providers repo, along with the corresponding\nimage configuration.\n\nAs far as we know, there should be no puppet or ansible relying on\nany of this.\n\nChange-Id: I149a0d146920424f6f4d202e5d915406b92e9bb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7ed9a22e97183022886c7c5b0b2b7c61ddf27fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b7ed9a22e97183022886c7c5b0b2b7c61ddf27fb"}]},"branch":"refs/heads/master"},"94b1363f4cb9e94aabb5a18af74584d5a84c28ee":{"kind":"REWORK","_number":2,"created":"2025-08-06 16:42:42.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/84/956184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/956184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/956184/2"}}},"commit":{"parents":[{"commit":"9d2c590bb986ebcf29ef74593a7cf32f55bd82c3","subject":"Merge \"Remove CLA enforcement from all projects and lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d2c590bb986ebcf29ef74593a7cf32f55bd82c3"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-30 18:01:19.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-08-06 16:42:36.000000000","tz":-420},"subject":"Remove nodepool configuration/elements","message":"Remove nodepool configuration/elements\n\nThis removes the nodepool configuration and DIB elements.  Our use\nof nodepool has been replaced by zuul-launcher, and our elements have\nmoved to the zuul-providers repo, along with the corresponding\nimage configuration.\n\nAs far as we know, there should be no puppet or ansible relying on\nany of this.\n\nChange-Id: I149a0d146920424f6f4d202e5d915406b92e9bb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94b1363f4cb9e94aabb5a18af74584d5a84c28ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94b1363f4cb9e94aabb5a18af74584d5a84c28ee"}]},"branch":"refs/heads/master"},"b22bdcf480bf91562709da61252f14ba3c2f6f46":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-06 19:38:43.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/84/956184/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/956184/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/956184/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/956184/3"}}},"commit":{"parents":[{"commit":"fd40838964afceef90d91255302b4d624df30c1d","subject":"Stop running project-config-dib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd40838964afceef90d91255302b4d624df30c1d"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-07-30 18:01:19.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-08-06 19:38:33.000000000","tz":-420},"subject":"Remove nodepool configuration/elements","message":"Remove nodepool configuration/elements\n\nThis removes the nodepool configuration and DIB elements.  Our use\nof nodepool has been replaced by zuul-launcher, and our elements have\nmoved to the zuul-providers repo, along with the corresponding\nimage configuration.\n\nAs far as we know, there should be no puppet or ansible relying on\nany of this.\n\nChange-Id: I149a0d146920424f6f4d202e5d915406b92e9bb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b22bdcf480bf91562709da61252f14ba3c2f6f46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b22bdcf480bf91562709da61252f14ba3c2f6f46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
