)]}'
{"id":"openstack%2Fproject-config~956726","triplet_id":"openstack%2Fproject-config~master~Ia97beb658ddec372eae9c86fa61e5bba92ca31d0","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-08-06 20:28:43.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia97beb658ddec372eae9c86fa61e5bba92ca31d0","subject":"Stop running project-config-dib","status":"MERGED","created":"2025-08-06 19:38:43.000000000","updated":"2025-08-06 20:37:51.000000000","submitted":"2025-08-06 20:35:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956726","meta_rev_id":"3d8653bedda33691aa5894cc2a7cbdcfaa954c3d","_number":956726,"virtual_id_number":956726,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 20:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2025-08-06 20:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2025-08-06 19:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2025-08-06 20:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 19:46:37.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-08-06 19:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecac4b1a0ba9174c4c84b3e0ffff50040ffbc47a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-08-06 19:38:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0f67b9d83abd3f6c5621c1d0a202bf8bfa5aa80","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-06 19:46:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"679c6fabc403db20dcc924c128a4f9fb110b39f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:52:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e200d21a44d4124b121862e185d6c98\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2b3688c148234dd49ecb649c38ab5100 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"552020f10a18ca1abc8f9efb0522643ebf6b1a84","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-08-06 20:28:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3578fc3713dcc5974c13b5b29fa4636f990f6775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:29:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2c73629b83aee817387c8ce4247e42ad4437d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:35:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caccfa7ad5f74b46b3dd8bd9f55987dc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ae9d68a6a93143138b4489fc85a4be75 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ba0e7f43f5865c4a192135618ab4c289dd673b1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:35:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d8653bedda33691aa5894cc2a7cbdcfaa954c3d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:37:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8acdf0971d1e4b62b8d1f3361b552ef9\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/d2ba40f033fd4655a8eb475dbd4260d2 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd40838964afceef90d91255302b4d624df30c1d","revisions":{"fd40838964afceef90d91255302b4d624df30c1d":{"kind":"REWORK","_number":1,"created":"2025-08-06 19:38:43.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/26/956726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/956726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/956726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/956726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/956726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/956726/1"}}},"commit":{"parents":[{"commit":"abb63fbd5ad111ec659b793124731adbc14f6709","subject":"Merge \"Remove networking-midonet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abb63fbd5ad111ec659b793124731adbc14f6709"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-08-06 19:37:58.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-08-06 19:37:58.000000000","tz":-420},"subject":"Stop running project-config-dib","message":"Stop running project-config-dib\n\nWe\u0027re about to remove all of the DIB config from this repo; stop\nrunning the job that validates it.\n\nChange-Id: Ia97beb658ddec372eae9c86fa61e5bba92ca31d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd40838964afceef90d91255302b4d624df30c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd40838964afceef90d91255302b4d624df30c1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
