)]}'
{"id":"openstack%2Fproject-config~959574","triplet_id":"openstack%2Fproject-config~master~Ia783a22dbd9701ae802f4037729f013ebd308c0f","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"34817":{"account":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"last_update":"2025-09-04 19:27:33.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2025-09-04 19:27:33.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-09-04 19:23:22.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia783a22dbd9701ae802f4037729f013ebd308c0f","subject":"Add repo app-metallb for starlingx","status":"MERGED","created":"2025-09-04 18:14:03.000000000","updated":"2025-09-04 19:47:00.000000000","submitted":"2025-09-04 19:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959574-new-project","meta_rev_id":"bece3218ca6f793db345efa97083013b44c8a243","_number":959574,"virtual_id_number":959574,"owner":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 19:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-09-04 18:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 19:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 19:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 18:20:10.000000000","updated_by":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2025-09-04 18:20:10.000000000","updated_by":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-09-04 18:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-04 18:37:58.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"7508121b598b7100952be1c7089c074c87666292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"date":"2025-09-04 18:14:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63e2fe746280fd8e118230e8d7335da5002831a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"date":"2025-09-04 18:18:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d638280d3feda9b793a397f2334165125a9577cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:25:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81998749549447be9583ee950f57ca78\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ebbf8d0fc7a1427fba26fe7119b84ad3 : SUCCESS in 5m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5298dc3a38664b4a858d9ff852c43a75 : FAILURE in 2m 17s","accounts_in_message":[],"_revision_number":2},{"id":"73885eb9a381ca0b36dabc788f40d59f3b463565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"date":"2025-09-04 18:33:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"19b9a873d78c5e240981655e7d85cd69232d8697","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-09-04 18:37:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a50c25f2cf971f05e51530b89539fdfffbab5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:42:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e77088e821cb44c89fbcf060d4c1671a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/83f6e79832d742538a539be801321d79 : SUCCESS in 4m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bf977bc1f2a84616b89c9c92b64163d0 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a81d899e4a3723fce726ed7661fc805c747126e5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-09-04 19:23:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bf5a2b81e4342facc9ae2ed1681b43717ef8f558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:23:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dbb8db171da61bb4b355572155294fabdb4ffced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:27:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dd02dc1275d4e22ba2a4b3b5d5490d5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d7fbec54fbc84e08b091008bd43c9e99 : SUCCESS in 2m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6aec8986983243b78cbbfae8e572c5b8 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"2b1d7c3834509c05e3b4293df4316df430c7bf54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:27:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d8bb98bf69e82b560f9d6a3138ef66b9c10d0fb7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:42:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018f0728bacb42a5971716f1ad6e195c\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/da0411ed2df84b43b664c2e2731923bf : SUCCESS in 14m 32s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/52dc51a7c9554f0e9ae89d896c9c4087 : SUCCESS in 7m 29s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/d3d5227469b54f2fb8070bc581827f22 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"bece3218ca6f793db345efa97083013b44c8a243","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-09-04 19:47:00.000000000","message":"Patch Set 3:\n\nOops, I meant to get Scott Little to review this first, but anyway I\u0027ve added him as the initial member for the starlingx-app-metallb-core group now.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ce6015d1081f44d8c1c9aa014b772da1c68c69e7","revisions":{"448f200c9927f70bb45ac0a64a26bca14835d893":{"kind":"REWORK","_number":1,"created":"2025-09-04 18:14:03.000000000","uploader":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"ref":"refs/changes/74/959574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/959574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/959574/1"}}},"commit":{"parents":[{"commit":"506e05fde1c78460ab7b0d358865272031852b57","subject":"Stop using storyboard by Freezer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/506e05fde1c78460ab7b0d358865272031852b57"}]}],"author":{"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","date":"2025-09-04 18:12:46.000000000","tz":-240},"committer":{"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","date":"2025-09-04 18:12:46.000000000","tz":-240},"subject":"Add repo app-metallb for starlingx","message":"Add repo app-metallb for starlingx\n\nThis commit adds a new repo named app-metallb, a MetalLB application\nrepo for StarlingX\n\nChange-Id: Ia783a22dbd9701ae802f4037729f013ebd308c0f\nSigned-off-by: Mohammad Issa \u003cmohammad.issa@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/448f200c9927f70bb45ac0a64a26bca14835d893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/448f200c9927f70bb45ac0a64a26bca14835d893"}]},"branch":"refs/heads/master"},"38a52dfe259248218c17dce1e905f63fbc7a7709":{"kind":"REWORK","_number":2,"created":"2025-09-04 18:18:49.000000000","uploader":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"ref":"refs/changes/74/959574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/959574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/959574/2"}}},"commit":{"parents":[{"commit":"506e05fde1c78460ab7b0d358865272031852b57","subject":"Stop using storyboard by Freezer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/506e05fde1c78460ab7b0d358865272031852b57"}]}],"author":{"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","date":"2025-09-04 18:12:46.000000000","tz":-240},"committer":{"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","date":"2025-09-04 18:18:23.000000000","tz":-240},"subject":"Add repo app-metallb for starlingx","message":"Add repo app-metallb for starlingx\n\nThis commit adds a new repo named app-metallb, a MetalLB application\nrepo for StarlingX\n\nChange-Id: Ia783a22dbd9701ae802f4037729f013ebd308c0f\nSigned-off-by: Mohammad Issa \u003cmohammad.issa@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/38a52dfe259248218c17dce1e905f63fbc7a7709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/38a52dfe259248218c17dce1e905f63fbc7a7709"}]},"branch":"refs/heads/master"},"ce6015d1081f44d8c1c9aa014b772da1c68c69e7":{"kind":"REWORK","_number":3,"created":"2025-09-04 18:33:04.000000000","uploader":{"_account_id":34817,"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","username":"missa"},"ref":"refs/changes/74/959574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/959574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/959574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/959574/3"}}},"commit":{"parents":[{"commit":"506e05fde1c78460ab7b0d358865272031852b57","subject":"Stop using storyboard by Freezer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/506e05fde1c78460ab7b0d358865272031852b57"}]}],"author":{"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","date":"2025-09-04 18:12:46.000000000","tz":-240},"committer":{"name":"Mohammad Issa","email":"mohammad.issa@windriver.com","date":"2025-09-04 18:32:44.000000000","tz":-240},"subject":"Add repo app-metallb for starlingx","message":"Add repo app-metallb for starlingx\n\nThis commit adds a new repo named app-metallb, a MetalLB application\nrepo for StarlingX\n\nChange-Id: Ia783a22dbd9701ae802f4037729f013ebd308c0f\nSigned-off-by: Mohammad Issa \u003cmohammad.issa@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce6015d1081f44d8c1c9aa014b772da1c68c69e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ce6015d1081f44d8c1c9aa014b772da1c68c69e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
