)]}'
{"id":"openstack%2Fproject-config~961475","triplet_id":"openstack%2Fproject-config~master~Ib1e39307fbeeddff2b26a268fa56df700641439f","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-01-12 17:27:21.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-01-15 09:50:24.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-15 09:59:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1e39307fbeeddff2b26a268fa56df700641439f","subject":"Migrate base propose-updates job to Jammy","status":"MERGED","created":"2025-09-17 12:30:33.000000000","updated":"2026-01-15 10:02:30.000000000","submitted":"2026-01-15 09:59:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"961475","meta_rev_id":"b9830997f2482280956182b7ae7c2dda3a82d6f3","_number":961475,"virtual_id_number":961475,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-15 09:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-01-12 17:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2026-01-15 09:50:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-01-15 09:50:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 12:46:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 16:18:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-01-12 17:26:00.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-01-15 09:50:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"7fa7cf8d17389782040049c2328a853e388bc997","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-17 12:30:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06a7039ed9953c6ede3aaa5483bd20e2fcc0b20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 12:46:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d634428096454881e12ed9081d383f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f7667c94a4647bb9bf6d6ca4a3ae0c2 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"fef2f03ce3f2bd7c287e8de40bbbce07785dfb43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-17 12:46:51.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d764d1bec42a5ae09679d9041f9cf347c0f08a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 13:08:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3a8ec8b8cec40b5a1dee4cffd376251\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/574a11939f9c4bbb9e3341b2cbffa2fb : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9932cb91ad9a5b88bc8551cc31dad484568152a0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-12 16:18:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba13e457c8a0ebb74912e6eded45551322fc1ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-12 17:24:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4d2aa6a753a40e3106d4041f5c0e98fef141aedc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-01-12 17:26:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a49fab76fc7cfeab5a5ebdcbf93a3da662a968c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-12 17:26:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b4e6d1045ae82eb3a522bc534cb13e52b6c04c2e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-01-12 17:26:25.000000000","message":"Patch Set 3: -Workflow Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"59e241d2946787d4db5160462b3c24381e00f759","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-01-12 17:27:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9746d27e82c84edf4a2d6eb321d60ba088052974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 18:19:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa33904d3d7b46e9ae21a11e95e7f4cd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6b03168f038a4889abfaa8bc34da709a : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"3f5232855c00aab0686c0e82a4f6aa75bbdde485","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-15 09:50:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"413e1feb7ea30e20fbf01758c4f2cfec25e56d1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 09:50:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1d5250efe4ba0aac530d725bc4b8bb6a636efdc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 09:59:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/543f724fc3bd4ce98d362df51cca4d14\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/15211c7ef53e4d5c857acf5abb536790 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":4},{"id":"a09eebfba1ffb4c25125b3aa6ee2e71e801752d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 09:59:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b9830997f2482280956182b7ae7c2dda3a82d6f3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 10:02:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a38496547cb4611a9d39bfbfdaa3d5a\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/de872d42573b438a9cc93c96ed8a7a17 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"20eb396e91e41c1afb2a14d48389f3ea538e88a0","revisions":{"530348ed7f229d4d951720522a6c0aa897186ffd":{"kind":"REWORK","_number":1,"created":"2025-09-17 12:30:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/961475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/961475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/961475/1"}}},"commit":{"parents":[{"commit":"e6fcbe893988296e476091ead1a2224384c0c123","subject":"Merge \"Disable ruamel automagic line wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e6fcbe893988296e476091ead1a2224384c0c123"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:28:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:30:24.000000000","tz":540},"subject":"Migrate propose-updates job for p-o-i in noble","message":"Migrate propose-updates job for p-o-i in noble\n\n... to fix the git command currently failing due to the trailer option\nnot available.\n\nChange-Id: Ib1e39307fbeeddff2b26a268fa56df700641439f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/530348ed7f229d4d951720522a6c0aa897186ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/530348ed7f229d4d951720522a6c0aa897186ffd"}]},"branch":"refs/heads/master"},"6dfb391778e21a310d00b5cde1c17fabb978e836":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-17 12:46:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/961475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/961475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/961475/2"}}},"commit":{"parents":[{"commit":"e6fcbe893988296e476091ead1a2224384c0c123","subject":"Merge \"Disable ruamel automagic line wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e6fcbe893988296e476091ead1a2224384c0c123"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:28:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:46:51.000000000","tz":0},"subject":"Migrate propose-updates job for p-o-i to noble","message":"Migrate propose-updates job for p-o-i to noble\n\n... to fix the git command currently failing due to the trailer option\nnot available.\n\nChange-Id: Ib1e39307fbeeddff2b26a268fa56df700641439f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6dfb391778e21a310d00b5cde1c17fabb978e836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6dfb391778e21a310d00b5cde1c17fabb978e836"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b2603105082e8128576c9be14c8fc60594b54814":{"kind":"REWORK","_number":3,"created":"2026-01-12 17:24:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/961475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/961475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/961475/3"}}},"commit":{"parents":[{"commit":"16be43789218c6935c5fc4cabdea293849d224b2","subject":"Merge \"Add Kubernetes Module Manager app\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16be43789218c6935c5fc4cabdea293849d224b2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:28:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-12 17:23:50.000000000","tz":540},"subject":"Migrate base propose-updates job for p-o-i to Jammy","message":"Migrate base propose-updates job for p-o-i to Jammy\n\nSupport for python 3.8 and 3.9 was removed, and requirements update no\nlonger used these versions since [1] was merged.\n\nThis fixes the git command currently failing due to the trailer option,\nwhich was added by [2], not available in the git package in Focal.\n\n[1] 702037af47f00639a93566335029c159a5a33284\n[2] 9057be2b912c73bad34ca0914d88ee71f267fb75\n\nChange-Id: Ib1e39307fbeeddff2b26a268fa56df700641439f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2603105082e8128576c9be14c8fc60594b54814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2603105082e8128576c9be14c8fc60594b54814"}]},"branch":"refs/heads/master"},"20eb396e91e41c1afb2a14d48389f3ea538e88a0":{"kind":"REWORK","_number":4,"created":"2026-01-12 17:26:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/961475/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/961475/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/961475/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/961475/4"}}},"commit":{"parents":[{"commit":"16be43789218c6935c5fc4cabdea293849d224b2","subject":"Merge \"Add Kubernetes Module Manager app\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16be43789218c6935c5fc4cabdea293849d224b2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:28:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-12 17:25:46.000000000","tz":540},"subject":"Migrate base propose-updates job to Jammy","message":"Migrate base propose-updates job to Jammy\n\nSupport for python 3.8 and 3.9 was removed, and requirements update no\nlonger used these versions since [1] was merged.\n\nThis fixes the git command currently failing due to the trailer option,\nwhich was added by [2], not available in the git package in Focal.\n\n[1] 702037af47f00639a93566335029c159a5a33284\n[2] 9057be2b912c73bad34ca0914d88ee71f267fb75\n\nChange-Id: Ib1e39307fbeeddff2b26a268fa56df700641439f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20eb396e91e41c1afb2a14d48389f3ea538e88a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/20eb396e91e41c1afb2a14d48389f3ea538e88a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
