)]}'
{"id":"openstack%2Fproject-config~962116","triplet_id":"openstack%2Fproject-config~master~I98cf4a6b9c5094059b0fe410e3086e262c1b19ba","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-09-27 08:05:48.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2026-05-05 16:11:29.000000000","reason":"Change was marked work in progress"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-05-05 16:11:29.000000000","reason":"Change was marked work in progress"}},"hashtags":["retire-refstack"],"change_id":"I98cf4a6b9c5094059b0fe410e3086e262c1b19ba","subject":"Remove refstack projects","status":"NEW","created":"2025-09-23 15:08:13.000000000","updated":"2026-05-12 18:19:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"7a7826f6ac38804708ce0bbb28114a9ace10f213","_number":962116,"virtual_id_number":962116,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-12 18:19:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-09-23 15:08:33.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-09-23 15:08:33.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-09-23 15:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eed89dd99115a9944d24e29c17eee2c48f992ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2025-09-23 15:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e324ba14e1e4f1ec3bc1c1d912832d2117183c5","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2025-09-23 15:10:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d405f1339a19011534b857e31d46f02243434593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 15:16:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/674f7941fddd4826bd78ce008f2d1214\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/78e49abdc99349fa89a851c072bb6b0f : SUCCESS in 4m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/44a60f964f0f4a009130bf4ed8cd794c : FAILURE in 3m 14s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/138036e1b5024635b469a0ecee256233 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d1965284a0bcc72fb4edbc5ed1773d613e1be53c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2025-09-23 18:44:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"65c671c8cd8f1284b61d3143ec3c1de8f1e00026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 18:45:26.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"bff92680ddd53d8d44282454a377decc601c14cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2025-09-23 18:52:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"14f478b53203eeba47a0dd5c0896e1ecbc88e3c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 18:53:16.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"43240049f67d4221d9bad5f0c3b0163919be4e03","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-09-23 19:22:57.000000000","message":"Hashtag added: retire-refstack","accounts_in_message":[],"_revision_number":3},{"id":"d24e0eb29a74640e6b932517c1e9111f72d544e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 16:11:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nThe following files contain Git conflicts:\n* gerritbot/channels.yaml\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ef69c7629c4a17b55aac3671605e518ac5ec98ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 16:12:07.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"8cc6d31284976a174ffc583c6e72bf30d49cf044","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 16:12:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c7053daedfc27faa5f2e45da6c60d475e9ce9dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:26:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b232dfddb4b447a398c9746b8b45026f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea589ca1635f451d94db57e01fedb6fa : SUCCESS in 4m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5977d91cfa804babb3c46a3b1a14f14f : FAILURE in 3m 21s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7740d6c1b5fb48d49158111ba4099ab1 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":6},{"id":"3d6a696afcdf9b5f46a096d4cecf92d2e1cf3cf4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 16:30:37.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"6d0cee9b5694f532e484692bce966e8e0074f199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:41:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e8a96ad25164186a5efc4227f2b6fa7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/16c2783496cd4dc1b6dc61267dcc353b : SUCCESS in 2m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/63a57a96c1d243b7a9a2c042883e688d : FAILURE in 3m 14s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/80eff0f2416146aea5743a923cc13d84 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":7},{"id":"3668dcef7ac9bf806b526116de1df5af7ff04784","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-07 17:30:32.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"1b4ae356558a139217f528258a1804de3a1634ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 17:37:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20cbcba6c00e4434b4819978d0ba1d2a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f7b42b92fc384254803653332c721e18 : SUCCESS in 3m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f5b8088e78ad4bd9a07499551e3e6ae7 : SUCCESS in 3m 37s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/6a3471d8d873469d9a831719f3e70bd4 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":8},{"id":"6f681299c3a2efc81f66cd243ad06ed761bb1e05","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-07 19:41:05.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"74959cd3e1a184841eea8775fc1d73dab98ce5bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 19:52:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b144a4cace144168f0fc0a665577851\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de0c7c12f21d4eeabf2907d61e395374 : SUCCESS in 4m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8328d52381a047f3b8b12435c3b5fb55 : SUCCESS in 2m 27s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/045515cdef4d48f08caa4ca0eaba3d26 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":9},{"id":"0066eaa6213276bd93b2d9857a9c051ca7ee0173","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-12 18:05:55.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"6ffe396aa81980bcb04378714f48b6577603e904","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-12 18:06:58.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"7a7826f6ac38804708ce0bbb28114a9ace10f213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:19:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dad51e79f9724412b488073c0849ca10\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5a5691ce7b844422992a646041170f06 : SUCCESS in 4m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/db24e56b53a34b649a4f95ce8eacf8aa : FAILURE in 3m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c8a195a7d58e4cbd9acbc0ba866efcab : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"f6cd99718a9494a4ed6144d3d0ad2fcfe5ad2da8","revisions":{"94c59546b3a0000b5059dc7e54b5b3e3b007ee9b":{"kind":"REWORK","_number":1,"created":"2025-09-23 15:08:13.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/16/962116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/1"}}},"commit":{"parents":[{"commit":"b2436928b69fd411d36a15edf8c83c2941b148ee","subject":"Fix update_constraints.sh for \u0027python_version\u003e\u003d\u0027 cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2436928b69fd411d36a15edf8c83c2941b148ee"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:08:11.000000000","tz":120},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94c59546b3a0000b5059dc7e54b5b3e3b007ee9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94c59546b3a0000b5059dc7e54b5b3e3b007ee9b"}]},"branch":"refs/heads/master"},"bda66873c441ddc2b4cc9f0b5d9dc408154e00b9":{"kind":"REWORK","_number":2,"created":"2025-09-23 18:44:01.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/16/962116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/2"}}},"commit":{"parents":[{"commit":"b2436928b69fd411d36a15edf8c83c2941b148ee","subject":"Fix update_constraints.sh for \u0027python_version\u003e\u003d\u0027 cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2436928b69fd411d36a15edf8c83c2941b148ee"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 18:43:41.000000000","tz":120},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bda66873c441ddc2b4cc9f0b5d9dc408154e00b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bda66873c441ddc2b4cc9f0b5d9dc408154e00b9"}]},"branch":"refs/heads/master"},"f6cc19e60597fe3e0864fb20a990e9515043d177":{"kind":"REWORK","_number":3,"created":"2025-09-23 18:52:27.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/16/962116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/3"}}},"commit":{"parents":[{"commit":"b2436928b69fd411d36a15edf8c83c2941b148ee","subject":"Fix update_constraints.sh for \u0027python_version\u003e\u003d\u0027 cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2436928b69fd411d36a15edf8c83c2941b148ee"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 18:48:57.000000000","tz":120},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6cc19e60597fe3e0864fb20a990e9515043d177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6cc19e60597fe3e0864fb20a990e9515043d177"}]},"branch":"refs/heads/master"},"ac5ab2156c0e28e4a5cf282111f1ac40b4996926":{"kind":"REWORK","_number":4,"created":"2026-05-05 16:11:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/4"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-05 16:11:29.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac5ab2156c0e28e4a5cf282111f1ac40b4996926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac5ab2156c0e28e4a5cf282111f1ac40b4996926"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"f6cc19e60597fe3e0864fb20a990e9515043d177","theirs":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"9bf9e7305940793d3f3328ef962a93f7fcce7a86":{"kind":"REWORK","_number":5,"created":"2026-05-05 16:12:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/5"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-05 16:12:06.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9bf9e7305940793d3f3328ef962a93f7fcce7a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9bf9e7305940793d3f3328ef962a93f7fcce7a86"}]},"branch":"refs/heads/master"},"6209013f17fb6e5ebbe4d42a6e7a3436b8ebbb3e":{"kind":"REWORK","_number":6,"created":"2026-05-05 16:12:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/6"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-05 16:12:32.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6209013f17fb6e5ebbe4d42a6e7a3436b8ebbb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6209013f17fb6e5ebbe4d42a6e7a3436b8ebbb3e"}]},"branch":"refs/heads/master"},"2dfec223ec6214dabebf9c4a28316e8b8e189a6c":{"kind":"REWORK","_number":7,"created":"2026-05-05 16:30:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/7"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-05 16:30:37.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2dfec223ec6214dabebf9c4a28316e8b8e189a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2dfec223ec6214dabebf9c4a28316e8b8e189a6c"}]},"branch":"refs/heads/master"},"929a0f1c3f646c26b2f518f03e802f0a36ee0315":{"kind":"REWORK","_number":8,"created":"2026-05-07 17:30:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/8"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-07 17:30:31.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/929a0f1c3f646c26b2f518f03e802f0a36ee0315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/929a0f1c3f646c26b2f518f03e802f0a36ee0315"}]},"branch":"refs/heads/master"},"3b3ce86cbaa7d8e213066a8407b70b6bfee3e63f":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2026-05-07 19:41:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/9"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-07 19:41:05.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nNo removal of refstack acl, because python-tempestconf\nis still using it\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b3ce86cbaa7d8e213066a8407b70b6bfee3e63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b3ce86cbaa7d8e213066a8407b70b6bfee3e63f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b1d8a35629efbc5e2c314243c6a6e1d3b3b51325":{"kind":"REWORK","_number":10,"created":"2026-05-12 18:05:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/10"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-12 18:05:54.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nNo removal of refstack acl, because python-tempestconf\nis still using it\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1d8a35629efbc5e2c314243c6a6e1d3b3b51325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1d8a35629efbc5e2c314243c6a6e1d3b3b51325"}]},"branch":"refs/heads/master"},"f6cd99718a9494a4ed6144d3d0ad2fcfe5ad2da8":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-05-12 18:06:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/962116/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/962116/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/962116/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/962116/11"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-09-23 15:04:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-12 18:06:58.000000000","tz":0},"subject":"Remove refstack projects","message":"Remove refstack projects\n\nNew Gerrit groups for management of python-tempestconf\nwill be created and populated with existing members\nof refstack groups\n\nDepends-on: I688aecc8ff228936217be0aba9b5fb163d38de7d\nDepends-on: I85c44dc83727d035e95931f2e7528aaf5772b314\nDepends-on: I849fe90d952ac2b5696e4eb90a48d69db59ec51d\n\nChange-Id: I98cf4a6b9c5094059b0fe410e3086e262c1b19ba\nSigned-off-by: Christian Berendt \u003cberendt@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6cd99718a9494a4ed6144d3d0ad2fcfe5ad2da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6cd99718a9494a4ed6144d3d0ad2fcfe5ad2da8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
