)]}'
{"id":"openstack%2Fproject-config~962557","triplet_id":"openstack%2Fproject-config~master~I8756308ebfd1b54266d5540a6345d69273c6ccb6","project":"openstack/project-config","branch":"master","topic":"project-yaml-normalization","hashtags":[],"change_id":"I8756308ebfd1b54266d5540a6345d69273c6ccb6","subject":"Normalize projects.yaml","status":"MERGED","created":"2025-09-30 02:17:27.000000000","updated":"2025-10-20 06:41:57.000000000","submitted":"2025-10-20 06:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962557-project-yaml-normalization","meta_rev_id":"643c4e41146d26f0f9b1428ae7d7906e34eae432","_number":962557,"virtual_id_number":962557,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-20 06:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-10-20 06:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-10-20 06:24:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 02:29:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-20 06:24:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7d10f6469a3e71be84319218fd37dc5b327e564f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-30 02:17:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baab0de5f077b03948c5ba721b2d23e06d7c8c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 02:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93866c75d2934432aa640857a59bd314\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ae9a5592daed4646a2df19aed0316e68 : SUCCESS in 4m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a62cf49cf9ce41bf84560a66884f8683 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7a9dbc32d0b223db6bdfc5dd2c6b0db16d7b41ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-01 02:17:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e37ad3bfba2935706451be811c38d45361a2acc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 02:33:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bcf05e9a0aa45a7b96608f08f87a790\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1af07ca4e2cd4659adb22116287ee06e : SUCCESS in 6m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5829ee3c4517475bbd14660eee90afed : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7d476b224ef0115c04cf8a5c9c7ba80297495688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-02 02:13:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c1a9d0a3f73cd9428989c68703e3052786a6210b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 02:37:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/430c7036699f47c18dd9ba62eb74c736\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/21181c8e0c0e4fc28d7db6fdd285908b : SUCCESS in 4m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/de737ea4f5c84543b4d8c02979fa7432 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e2d9f0e56e061b586fa83f35345873404d463f50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-03 02:12:29.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"16dc533c177edfc02d10b20e9684d94fd5c097fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 02:24:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2643d0bb42c54881a670355e84a14f52\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/496a9c4a39ac472496eae4999352d918 : SUCCESS in 3m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f6ab56917a294edf8450e45e981f5a2c : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"ecf9c160cfed69e5bbce22706a49f225c921b943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-05 02:20:45.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5d54307140b6c043d06870c0610a78b7ab2720f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-05 02:56:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/370f6b0c40af4452a007cd3e06032b2a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9b9305271cf94c1b9bc4056e3eed844f : SUCCESS in 5m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fb834377bb25412aacb0d11dfac15e26 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":5},{"id":"42019dd29743a8f37164e75718a9a646fd9c3cdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-06 02:15:38.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1d8b78869fddabcec90aeffe2acdc57e2aac9476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 02:29:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55d4cd4ff2f340328e68464c5e254a20\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9bc21c09036f49288261fc803e31a831 : SUCCESS in 3m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/898bb564e58b4e8fa7b23397ef063377 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"ea131481a13a7ff0ab6d7a9df6b23a480332069a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-07 02:25:01.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b434899ba93ea23c59f3bc796262e26bcac86fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 02:32:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a517e5127f474c9c9449e6bb1a69bc54\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3292419308b94da09edc1849fe35d4f8 : SUCCESS in 4m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/de9879e9c61544ada426e1e99bd20617 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":7},{"id":"42cd5f3f8d7fefbbf2d571d86be17536b9daa0d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-08 02:24:30.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"951bb319a00cc9d05d58ca346fb7b9b1534bd805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 02:44:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39badcd75ee440bcaf43ff676fc7c15e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/87ea5e67627d4649be16360e50cfd3e4 : SUCCESS in 4m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d25978dc5dd44d36b7484daca2acd708 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":8},{"id":"2c1f76839673e28b6b04bfe81bea89c44ed7e691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-09 02:24:08.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"9a96337196680c9c09a3a561e6bcc34b88c754c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 02:47:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ca34480fc140509b3649d4dc64e547\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea776c1594af429e98f8ddf029785887 : SUCCESS in 5m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fdd3e8ac8f2742a69fad84e112816891 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":9},{"id":"394d18d6dfa413888fda501aed5e614039add1ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-10 02:19:29.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent tree, and commit message as Patch Set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"ad97e3d5768654c5c9bab84a530b1bf64c0b1aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 02:42:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d8ab4f8eca0486b95e1370195715464\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a884d8cd529e47f1b949f788ce764452 : SUCCESS in 5m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d05cafa67e4e4637ae4220635c6b45c0 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":10},{"id":"8d90873613699e78d8f22ec49ca3e3bd5e4cd2d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-11 02:30:05.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent tree, and commit message as Patch Set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"62c5ebd64760ea85ec1d25d7da0b3b50fa035a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 02:40:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc50d3c1933a4292b94740a14708652f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85803557f3ee4566be2bc5e04ab5742b : SUCCESS in 4m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/193dfa5179c4493c853bcaffc04c94b2 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":11},{"id":"eb3c0e8b2948f65152f1851e6de7a92956c69b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-12 02:22:13.000000000","message":"Uploaded patch set 12: New patch set was added with same tree, parent tree, and commit message as Patch Set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"5a84552af2e0874b95b69f03a9fba4333b7e1534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-12 02:37:56.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1486d206b52485daf3c1e94772ebc7d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/50ba3c71792d465aa60ce4ebd77ec13b : SUCCESS in 4m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/69e01296cda64ad68dcd6a78328a2ac0 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":12},{"id":"545c4473adaad7c4ffdde1f9da3bd81a02902639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-13 02:29:34.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"af44dde6c7db411ce3311588bec6e0522ce1b8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 02:41:41.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7afc7883ad9e4a2e8d4d0e02d3905400\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1cffb17ff2654ec0a43065bca2e44c6d : RETRY_LIMIT in 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/020f2dcb7d78424bb2f39dc700fe6008 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":13},{"id":"071dcf0b65f406c24da1bfcd929822e7c4d5c8dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-14 02:22:20.000000000","message":"Uploaded patch set 14: New patch set was added with same tree, parent tree, and commit message as Patch Set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"7a26b99e2881c82cbef39bc925398e938eef1824","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:32:28.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5de931d6d91476dbc981543d336227f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/89fcdfd9e9d14a3396c043c14107bc46 : SUCCESS in 5m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9e45121788194fedbedb7c0dc2f26ff8 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":14},{"id":"6a4fcbc3be115c8bd16b123a9acdefd0051accd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-15 02:15:48.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"ce38d78dc59ccdcf34670db9b6c7f63ff5dd8825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 02:44:08.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24e61921bd2e4ec1bcc519f804f6d6f8\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/593c029031a449b68ab8c3261d226d40 : SUCCESS in 3m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2226a5c24ef046c1987d53be7e6f246f : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":15},{"id":"d32f177f765c2f14307485cfc8a3621bef445f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-16 02:14:59.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"516a603b864723fa092b72f5e2090548d1303381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 02:29:08.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4406f9858d3486da62f0c6bb2e52326\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/31b6684eb03747758eee74baa7d11242 : SUCCESS in 5m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/584e343c37e8475fb03eed476f6728ca : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":16},{"id":"ee0d585f2ea7be9c93bbaf3d567038b52592a91f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-17 02:25:48.000000000","message":"Uploaded patch set 17: New patch set was added with same tree, parent tree, and commit message as Patch Set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"3a563f1d52774518cc9e0d1f849ac901f653d00c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 02:50:37.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd4550aa68e4ed699bd351084506fb1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fc855457f0034603882f7befac20b992 : SUCCESS in 4m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2dbf70831c1d49a7a19347b2717a0a83 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":17},{"id":"91350053c3f9499f5c1b070e6b31b359d50f2b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-18 02:21:05.000000000","message":"Uploaded patch set 18: New patch set was added with same tree, parent tree, and commit message as Patch Set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"30a56193956e9bddb676ae0b1a15147e041a5e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 02:31:41.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a19176bc5d8436b9bec6e2b88871345\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/069a90de9382449d8ad32889da223756 : SUCCESS in 3m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/20b9f17359cd4f55943a4520fca1dfc4 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":18},{"id":"911ad615291088d3ce5c92016fda6c14e8f810bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-19 02:15:55.000000000","message":"Uploaded patch set 19: New patch set was added with same tree, parent tree, and commit message as Patch Set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"053ade282a408f0cc60322bbc2bf033355447012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-19 02:37:00.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/615ca95f32464197aedbb65c0ef34bad\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/28faec49ea6042fba0ba45d31afe06bd : SUCCESS in 5m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3598e1fba6b14dc28f2d762e3472ff06 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":19},{"id":"f394682df3b4140282630126fd5e0bd569f3be30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-20 02:18:22.000000000","message":"Uploaded patch set 20: New patch set was added with same tree, parent tree, and commit message as Patch Set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"493f8cb8a129666971abc459e9ea2ab9eaeb21f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 02:32:25.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82aae613f6f14c9c94f1a429af9bfb2f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8aef01acfcda4f36bde2aeba7e218c90 : SUCCESS in 4m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0e725151bd194480bd30c6140ea60241 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":20},{"id":"3cda395be521df956811d19a32edd53fdaf8f10a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-10-20 06:24:22.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"f5f8435d88552d7e1f3a8365d4d53b15e9c07d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 06:24:31.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"9300b1ad9dbac9a4b092db426f4c285819d308b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 06:33:36.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff0f4f4bed043d98ddc06fa505d8187\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7c3203320cfb4ff6806ea25cfe500f76 : SUCCESS in 4m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f09d586f4c21448fac06df9f032eeabd : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":20},{"id":"719f2dfd9b641d01c687fda39a23887aa8f42417","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 06:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"643c4e41146d26f0f9b1428ae7d7906e34eae432","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 06:41:57.000000000","message":"Patch Set 20:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde137152b4a4f9e91a647b38b701c80\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/55eb0b9970a747fe820a9a67a6b88809 : SUCCESS in 8m 04s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/cf13dc361aae46258a74baed6038996b : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"cb01c1a4ddd780adef3b3721456a7aea094a5140","revisions":{"8bd6bf3d5653dbff365bf364c75b7763a7433972":{"kind":"REWORK","_number":1,"created":"2025-09-30 02:17:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/1"}}},"commit":{"parents":[{"commit":"169edacafa3b9b0aa12c3cb2135f5a365e94c60d","subject":"Retire shade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/169edacafa3b9b0aa12c3cb2135f5a365e94c60d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-30 02:17:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-30 02:17:24.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8bd6bf3d5653dbff365bf364c75b7763a7433972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8bd6bf3d5653dbff365bf364c75b7763a7433972"}]},"branch":"refs/heads/master"},"8caf8285ea52894ef65c2ba8ff92aca31f93b5bf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-01 02:17:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/2"}}},"commit":{"parents":[{"commit":"b5915bd52aa7791801ae3c4b3f6db648e12b1b13","subject":"Merge \"Temporarily remove release docs semaphores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5915bd52aa7791801ae3c4b3f6db648e12b1b13"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-01 02:17:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-01 02:17:38.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8caf8285ea52894ef65c2ba8ff92aca31f93b5bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8caf8285ea52894ef65c2ba8ff92aca31f93b5bf"}]},"branch":"refs/heads/master"},"5f61cbbf24cc29164fd1014db1b8811e5fa76bb6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-02 02:13:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/3"}}},"commit":{"parents":[{"commit":"5e591a28feee3eaf4bbb3c76512255f43ffc69c2","subject":"Merge \"Revert \"Temporarily remove release docs semaphores\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e591a28feee3eaf4bbb3c76512255f43ffc69c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-02 02:13:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-02 02:13:15.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f61cbbf24cc29164fd1014db1b8811e5fa76bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f61cbbf24cc29164fd1014db1b8811e5fa76bb6"}]},"branch":"refs/heads/master"},"37ce6ff7eff5b3d0e97bc5612544088be54ab054":{"kind":"NO_CHANGE","_number":4,"created":"2025-10-03 02:12:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/4"}}},"commit":{"parents":[{"commit":"5e591a28feee3eaf4bbb3c76512255f43ffc69c2","subject":"Merge \"Revert \"Temporarily remove release docs semaphores\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e591a28feee3eaf4bbb3c76512255f43ffc69c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-03 02:12:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-03 02:12:27.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37ce6ff7eff5b3d0e97bc5612544088be54ab054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37ce6ff7eff5b3d0e97bc5612544088be54ab054"}]},"branch":"refs/heads/master"},"d55dec92fd3103453c0c60a762b6df6cc2b38d7d":{"kind":"NO_CHANGE","_number":5,"created":"2025-10-05 02:20:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/5"}}},"commit":{"parents":[{"commit":"5e591a28feee3eaf4bbb3c76512255f43ffc69c2","subject":"Merge \"Revert \"Temporarily remove release docs semaphores\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e591a28feee3eaf4bbb3c76512255f43ffc69c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-05 02:20:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-05 02:20:43.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d55dec92fd3103453c0c60a762b6df6cc2b38d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d55dec92fd3103453c0c60a762b6df6cc2b38d7d"}]},"branch":"refs/heads/master"},"8c8875e1663639181b5c2acc1b03f631984900d2":{"kind":"NO_CHANGE","_number":6,"created":"2025-10-06 02:15:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/6"}}},"commit":{"parents":[{"commit":"5e591a28feee3eaf4bbb3c76512255f43ffc69c2","subject":"Merge \"Revert \"Temporarily remove release docs semaphores\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e591a28feee3eaf4bbb3c76512255f43ffc69c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-06 02:15:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-06 02:15:36.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c8875e1663639181b5c2acc1b03f631984900d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c8875e1663639181b5c2acc1b03f631984900d2"}]},"branch":"refs/heads/master"},"7cb0f832d13df1300c174f068352b4432e98cfdd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-07 02:25:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/7"}}},"commit":{"parents":[{"commit":"ddffc7a0d90cd39a2ac2025d7432d30d80fd1508","subject":"Merge \"Replace 2025.2/Flamingo key with 2026.1/Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ddffc7a0d90cd39a2ac2025d7432d30d80fd1508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-07 02:25:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-07 02:25:00.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7cb0f832d13df1300c174f068352b4432e98cfdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7cb0f832d13df1300c174f068352b4432e98cfdd"}]},"branch":"refs/heads/master"},"53058c284343ea3a0839dd23fb93780dbe71c684":{"kind":"NO_CHANGE","_number":8,"created":"2025-10-08 02:24:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/8"}}},"commit":{"parents":[{"commit":"ddffc7a0d90cd39a2ac2025d7432d30d80fd1508","subject":"Merge \"Replace 2025.2/Flamingo key with 2026.1/Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ddffc7a0d90cd39a2ac2025d7432d30d80fd1508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-08 02:24:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-08 02:24:28.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53058c284343ea3a0839dd23fb93780dbe71c684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53058c284343ea3a0839dd23fb93780dbe71c684"}]},"branch":"refs/heads/master"},"4bb71e55ff52bc578aeecf6f7e4afa7c4cc9475c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-10-09 02:24:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/9"}}},"commit":{"parents":[{"commit":"503e6d2735ae681c783fb8a1bddefbf487797430","subject":"ironic-inspector retirement: end most gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/503e6d2735ae681c783fb8a1bddefbf487797430"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-09 02:24:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-09 02:24:07.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bb71e55ff52bc578aeecf6f7e4afa7c4cc9475c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bb71e55ff52bc578aeecf6f7e4afa7c4cc9475c"}]},"branch":"refs/heads/master"},"09b8d9ccb186ff565aea6ea0a8cebe5d786227a2":{"kind":"NO_CHANGE","_number":10,"created":"2025-10-10 02:19:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/10"}}},"commit":{"parents":[{"commit":"503e6d2735ae681c783fb8a1bddefbf487797430","subject":"ironic-inspector retirement: end most gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/503e6d2735ae681c783fb8a1bddefbf487797430"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-10 02:19:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-10 02:19:27.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/09b8d9ccb186ff565aea6ea0a8cebe5d786227a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/09b8d9ccb186ff565aea6ea0a8cebe5d786227a2"}]},"branch":"refs/heads/master"},"930cd330276241339fc08b0764a5f5a22efcef4a":{"kind":"NO_CHANGE","_number":11,"created":"2025-10-11 02:30:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/11"}}},"commit":{"parents":[{"commit":"503e6d2735ae681c783fb8a1bddefbf487797430","subject":"ironic-inspector retirement: end most gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/503e6d2735ae681c783fb8a1bddefbf487797430"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-11 02:30:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-11 02:30:04.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/930cd330276241339fc08b0764a5f5a22efcef4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/930cd330276241339fc08b0764a5f5a22efcef4a"}]},"branch":"refs/heads/master"},"37e560abd4209dcb25820e1fbdd75f896fb4148a":{"kind":"NO_CHANGE","_number":12,"created":"2025-10-12 02:22:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/12"}}},"commit":{"parents":[{"commit":"503e6d2735ae681c783fb8a1bddefbf487797430","subject":"ironic-inspector retirement: end most gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/503e6d2735ae681c783fb8a1bddefbf487797430"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-12 02:22:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-12 02:22:12.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37e560abd4209dcb25820e1fbdd75f896fb4148a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37e560abd4209dcb25820e1fbdd75f896fb4148a"}]},"branch":"refs/heads/master"},"b8991c3b711f1105befda2c173b166aa76855799":{"kind":"NO_CHANGE","_number":13,"created":"2025-10-13 02:29:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/13"}}},"commit":{"parents":[{"commit":"503e6d2735ae681c783fb8a1bddefbf487797430","subject":"ironic-inspector retirement: end most gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/503e6d2735ae681c783fb8a1bddefbf487797430"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-13 02:29:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-13 02:29:32.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b8991c3b711f1105befda2c173b166aa76855799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b8991c3b711f1105befda2c173b166aa76855799"}]},"branch":"refs/heads/master"},"4713fd5296fee4c4cbb295067db4a0b35a0822a8":{"kind":"NO_CHANGE","_number":14,"created":"2025-10-14 02:22:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/14"}}},"commit":{"parents":[{"commit":"503e6d2735ae681c783fb8a1bddefbf487797430","subject":"ironic-inspector retirement: end most gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/503e6d2735ae681c783fb8a1bddefbf487797430"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-14 02:22:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-14 02:22:16.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4713fd5296fee4c4cbb295067db4a0b35a0822a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4713fd5296fee4c4cbb295067db4a0b35a0822a8"}]},"branch":"refs/heads/master"},"5c12920e539b409fafadd78245547e0be3f2c35b":{"kind":"NO_CHANGE","_number":15,"created":"2025-10-15 02:15:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/15"}}},"commit":{"parents":[{"commit":"503e6d2735ae681c783fb8a1bddefbf487797430","subject":"ironic-inspector retirement: end most gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/503e6d2735ae681c783fb8a1bddefbf487797430"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-15 02:15:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-15 02:15:46.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c12920e539b409fafadd78245547e0be3f2c35b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c12920e539b409fafadd78245547e0be3f2c35b"}]},"branch":"refs/heads/master"},"743ca3c2d70241443d2bb29afcc35012577b6d8c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-10-16 02:14:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/16"}}},"commit":{"parents":[{"commit":"3c61bca2c4c4e490af640449fc03764d04ba55c6","subject":"Merge \"Have more files trigger test-release-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c61bca2c4c4e490af640449fc03764d04ba55c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-16 02:14:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-16 02:14:57.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/743ca3c2d70241443d2bb29afcc35012577b6d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/743ca3c2d70241443d2bb29afcc35012577b6d8c"}]},"branch":"refs/heads/master"},"42d285d258766e30953427bc3c6908ddbc176164":{"kind":"NO_CHANGE","_number":17,"created":"2025-10-17 02:25:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/17"}}},"commit":{"parents":[{"commit":"3c61bca2c4c4e490af640449fc03764d04ba55c6","subject":"Merge \"Have more files trigger test-release-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c61bca2c4c4e490af640449fc03764d04ba55c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-17 02:25:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-17 02:25:47.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/42d285d258766e30953427bc3c6908ddbc176164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/42d285d258766e30953427bc3c6908ddbc176164"}]},"branch":"refs/heads/master"},"b541521466b2de11d6adac80d0e713585bee667c":{"kind":"NO_CHANGE","_number":18,"created":"2025-10-18 02:21:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/18"}}},"commit":{"parents":[{"commit":"3c61bca2c4c4e490af640449fc03764d04ba55c6","subject":"Merge \"Have more files trigger test-release-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c61bca2c4c4e490af640449fc03764d04ba55c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 02:21:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 02:21:02.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b541521466b2de11d6adac80d0e713585bee667c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b541521466b2de11d6adac80d0e713585bee667c"}]},"branch":"refs/heads/master"},"9c401e4c94104119c46ee4eb04f831741307a062":{"kind":"NO_CHANGE","_number":19,"created":"2025-10-19 02:15:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/19"}}},"commit":{"parents":[{"commit":"3c61bca2c4c4e490af640449fc03764d04ba55c6","subject":"Merge \"Have more files trigger test-release-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c61bca2c4c4e490af640449fc03764d04ba55c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-19 02:15:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-19 02:15:53.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9c401e4c94104119c46ee4eb04f831741307a062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9c401e4c94104119c46ee4eb04f831741307a062"}]},"branch":"refs/heads/master"},"cb01c1a4ddd780adef3b3721456a7aea094a5140":{"kind":"NO_CHANGE","_number":20,"created":"2025-10-20 02:18:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/962557/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/962557/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/962557/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/962557/20"}}},"commit":{"parents":[{"commit":"3c61bca2c4c4e490af640449fc03764d04ba55c6","subject":"Merge \"Have more files trigger test-release-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c61bca2c4c4e490af640449fc03764d04ba55c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-20 02:18:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-20 02:18:20.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8756308ebfd1b54266d5540a6345d69273c6ccb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb01c1a4ddd780adef3b3721456a7aea094a5140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb01c1a4ddd780adef3b3721456a7aea094a5140"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
