)]}'
{"id":"openstack%2Fproject-config~965343","triplet_id":"openstack%2Fproject-config~master~Idcb14d147d75736be39b32cf80b1f1b2983512c4","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Idcb14d147d75736be39b32cf80b1f1b2983512c4","subject":"kolla: reorganise dashboards","status":"NEW","created":"2025-10-30 06:25:11.000000000","updated":"2025-10-30 16:46:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab44fc3651b0975e7c23cf1f70a7bdc3b9b929f1","_number":965343,"virtual_id_number":965343,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-30 16:15:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-10-30 16:46:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-30 06:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"766e2be12933087121f8c9a472256882f6ea244b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-30 06:25:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be12f2643aca25f282a04c0dd68fb94a2f6e9414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 06:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4d5aa71cbfa4fd198c18959bdd304b5\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f232584551d74c138f0c7d1d44779256 : SUCCESS in 25m 10s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c59b21bcf5574e3ebab9b6f88655a87c : SUCCESS in 3m 51s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/869fb30121af4f2097f04d4e0d869f56 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c27ac1e1f595c8a7c5cdb36ad2c04b706e52d074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-30 09:13:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"146c42bdd81d0c422d8438a96b32e18d90d2473c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:33:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc5a3302d84448a80c8f9214204bc33\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/10fa8299d043495bbc315c25a335f191 : SUCCESS in 15m 11s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/30fe118bf7bc4d7db0af59f09a3fd07c : SUCCESS in 2m 45s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/8d47b8d4a0274b1ab59cbaa841557bd7 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":2},{"id":"fe0bac17b36bd7e4308cd08c4c9a59e2ac18fbc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-30 15:48:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8e9ce54ff24d3829b1bc394a8f44d22651343eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 16:15:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175cea37a0914e27b904b89c835a49a4\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5829f6e8f9f1464b95c54f51af0b160a : SUCCESS in 21m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a16717e7bd5e4703a996a85a3b973f88 : SUCCESS in 3m 01s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/d069cb127f9d4a3db1591a3ffa504edc : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ab44fc3651b0975e7c23cf1f70a7bdc3b9b929f1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-30 16:46:14.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"52dbb332a8d41e82513b9f3506ec567c586f7a8a","revisions":{"efe7dd1801e732ff0aeb19fb9c17f5674ffd8eef":{"kind":"REWORK","_number":1,"created":"2025-10-30 06:25:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/43/965343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/965343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/965343/1"}}},"commit":{"parents":[{"commit":"cb01c1a4ddd780adef3b3721456a7aea094a5140","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb01c1a4ddd780adef3b3721456a7aea094a5140"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-30 06:25:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-30 06:25:09.000000000","tz":60},"subject":"kolla: Update dashboards after jobs rename","message":"kolla: Update dashboards after jobs rename\n\nChange-Id: Idcb14d147d75736be39b32cf80b1f1b2983512c4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efe7dd1801e732ff0aeb19fb9c17f5674ffd8eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efe7dd1801e732ff0aeb19fb9c17f5674ffd8eef"}]},"branch":"refs/heads/master"},"9b89dc1e02305ad355dc70e9da0ae6828d131dc2":{"kind":"REWORK","_number":2,"created":"2025-10-30 09:13:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/43/965343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/965343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/965343/2"}}},"commit":{"parents":[{"commit":"cb01c1a4ddd780adef3b3721456a7aea094a5140","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb01c1a4ddd780adef3b3721456a7aea094a5140"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-30 06:25:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-30 09:13:53.000000000","tz":-240},"subject":"kolla: reorganise dashboards","message":"kolla: reorganise dashboards\n\nSplit Kolla and Kolla-Ansible into separate dashboards,\nrename jobs because they have been renamed in Zuul configs\n\nChange-Id: Idcb14d147d75736be39b32cf80b1f1b2983512c4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b89dc1e02305ad355dc70e9da0ae6828d131dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b89dc1e02305ad355dc70e9da0ae6828d131dc2"}]},"branch":"refs/heads/master"},"52dbb332a8d41e82513b9f3506ec567c586f7a8a":{"kind":"REWORK","_number":3,"created":"2025-10-30 15:48:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/43/965343/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/965343/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/965343/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/965343/3"}}},"commit":{"parents":[{"commit":"cb01c1a4ddd780adef3b3721456a7aea094a5140","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb01c1a4ddd780adef3b3721456a7aea094a5140"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-30 06:25:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-30 15:48:35.000000000","tz":-240},"subject":"kolla: reorganise dashboards","message":"kolla: reorganise dashboards\n\nSplit Kolla and Kolla-Ansible into separate dashboards,\nrename jobs because they have been renamed in Zuul configs\n\nChange-Id: Idcb14d147d75736be39b32cf80b1f1b2983512c4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52dbb332a8d41e82513b9f3506ec567c586f7a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52dbb332a8d41e82513b9f3506ec567c586f7a8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
