)]}'
{"id":"openstack%2Fproject-config~965422","triplet_id":"openstack%2Fproject-config~master~Iaafba342816823dcbe32c816a15e99566fe873d7","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"35446":{"account":{"_account_id":35446,"name":"Igor Pires Soares","display_name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","username":"ipiresso"},"last_update":"2025-11-18 16:19:09.000000000","reason":"Change was submitted"},"37661":{"account":{"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"},"last_update":"2025-11-18 14:05:39.000000000","reason":"\u003cGERRIT_ACCOUNT_37661\u003e replied on the change","reason_account":{"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"}}},"hashtags":[],"change_id":"Iaafba342816823dcbe32c816a15e99566fe873d7","subject":"Add Kubernetes Module Manager app","status":"MERGED","created":"2025-10-30 18:47:05.000000000","updated":"2025-11-18 16:31:44.000000000","submitted":"2025-11-18 16:19:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965422-new-project","meta_rev_id":"0424ab8f54c0dd3d7028e5cc92a1c9bc43deaa4d","_number":965422,"virtual_id_number":965422,"owner":{"_account_id":35446,"name":"Igor Pires Soares","display_name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","username":"ipiresso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 16:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"},"all":[{"value":1,"date":"2025-11-18 16:02:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-18 14:05:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"},{"value":2,"date":"2025-11-18 16:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"},{"value":1,"date":"2025-11-18 16:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-30 18:52:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-17 18:28:09.000000000","updated_by":{"_account_id":35446,"name":"Igor Pires Soares","display_name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","username":"ipiresso"},"reviewer":{"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"},"state":"REVIEWER"},{"updated":"2025-11-18 16:02:24.000000000","updated_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2025-11-18 16:10:00.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"0b428dd545510058e952f6b20f88fb45ab1d8b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35446,"name":"Igor Pires Soares","display_name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","username":"ipiresso"},"date":"2025-10-30 18:47:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac0447f1b9f50223039f623b45b4545d06686772","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35446,"name":"Igor Pires Soares","display_name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","username":"ipiresso"},"date":"2025-10-30 18:49:03.000000000","message":"Topic “new-project” removed","accounts_in_message":[],"_revision_number":1},{"id":"3d02015783bdfcdce8895fd9347f3dd09845cca3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35446,"name":"Igor Pires Soares","display_name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","username":"ipiresso"},"date":"2025-10-30 18:49:14.000000000","message":"Topic set to new-project","accounts_in_message":[],"_revision_number":1},{"id":"070d8c65866aaf9290caeba7f1bd4f4005ceff94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:52:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c099668de44c72907c53a4ed689a4d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5bd5be12a02f4f8e95be88894515d94b : SUCCESS in 4m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7e8630e9ace84447802a3b07df03dc13 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"442e97d9c9c48f5d203bf6b7015487a5f6601350","author":{"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"},"date":"2025-11-18 14:05:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef278b8d1fa851eff5bdca2a529063a48482d9a7","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2025-11-18 16:02:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"646d9b6d199b770520818c1ecc8cc1849435c488","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-11-18 16:10:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0be1cab618690fc3668d31ac94b0301434f8c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 16:10:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f490a1d930e2c111a872e938ae30cbdf69e3082","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 16:19:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0676ff5bb044bbfbaa31118e6ecf0e5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6859114564d343d29a1fb69f12e74e25 : SUCCESS in 3m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4029f7d53f13406bbe490a5402665536 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7fe103acbe85544271650ff0fc9b1268fbc56048","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 16:19:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0424ab8f54c0dd3d7028e5cc92a1c9bc43deaa4d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 16:31:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11d9781f84934eecb08db2025841628f\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/07cd6e647e264bf682bddd236a84b735 : SUCCESS in 12m 20s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c7dfc02afaac47ae924fba05eaec5ad7 : SUCCESS in 5m 20s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/a55c585933c84042b017e7268f817e80 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e632160b41160f99bcdeb4fae81c592ece9be1fd","revisions":{"e632160b41160f99bcdeb4fae81c592ece9be1fd":{"kind":"REWORK","_number":1,"created":"2025-10-30 18:47:05.000000000","uploader":{"_account_id":35446,"name":"Igor Pires Soares","display_name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","username":"ipiresso"},"ref":"refs/changes/22/965422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/965422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/965422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/965422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/965422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/965422/1"}}},"commit":{"parents":[{"commit":"cb01c1a4ddd780adef3b3721456a7aea094a5140","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb01c1a4ddd780adef3b3721456a7aea094a5140"}]}],"author":{"name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","date":"2025-10-30 18:42:55.000000000","tz":-180},"committer":{"name":"Igor Soares","email":"Igor.PiresSoares@windriver.com","date":"2025-10-30 18:47:02.000000000","tz":-180},"subject":"Add Kubernetes Module Manager app","message":"Add Kubernetes Module Manager app\n\nThis commit adds Kubernetes Module Manager app repo to StarlingX.\n\nChange-Id: Iaafba342816823dcbe32c816a15e99566fe873d7\nSigned-off-by: Igor Soares \u003cIgor.PiresSoares@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e632160b41160f99bcdeb4fae81c592ece9be1fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e632160b41160f99bcdeb4fae81c592ece9be1fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37661,"name":"Eduardo Franca","display_name":"Eduardo França","email":"eduardo.dossantosfranca@windriver.com","username":"edossan1"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
