)]}'
{"id":"openstack%2Fproject-config~966981","triplet_id":"openstack%2Fproject-config~master~Ice255368db795897c3f620edbad35b57a3e7f2c2","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2025-11-13 10:59:30.000000000","reason":"Change was submitted"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-11-12 21:51:42.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-11-12 21:58:38.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ice255368db795897c3f620edbad35b57a3e7f2c2","subject":"Grafana: add core graphs to rackspace-flex","status":"MERGED","created":"2025-11-12 21:49:24.000000000","updated":"2025-11-13 11:03:10.000000000","submitted":"2025-11-13 10:59:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966981","meta_rev_id":"7e0ea48d64d54b472d365f19ca75dafaf202518f","_number":966981,"virtual_id_number":966981,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-13 10:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2025-11-12 21:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2025-11-12 21:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-12 21:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2025-11-12 21:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-12 21:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 21:49:55.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-11-12 21:50:30.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-11-12 22:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-13 10:16:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"ee32f32e6abd909502184e343acc25a15ca1ed03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-11-12 21:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6f12fd75491fb353ab0bed43b2f79e47d700f2c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-11-12 21:49:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1bd20e5969618c09a410fefcf9aee81e7b2bad2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-11-12 21:50:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a3870fb344098d2e75f3d3e3d8a999bbae40e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-11-12 21:50:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"acfada1e0be4dae0e5ec254b525b4af5e78aab63","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-11-12 21:51:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"78b77425f3041c400780cfe0bebb76004d757b81","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-11-12 21:58:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"903f8669a60c4db8c0378165a66cd0a6985e607d","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2025-11-12 21:58:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"204f81c839597f2b415050e9eb1ea119bbccc6cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 22:10:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d4342609ee745cf90002923f23cc2c4\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5ea1e475b0784584bbfff15acdd22971 : SUCCESS in 15m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1eda83cba9ab4a82bf4a7c5aa6084f91 : SUCCESS in 4m 29s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/90b592aab5574f4391771ded3126aefc : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":2},{"id":"eb28095add3d7e6e511911367865ee07428f4d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 22:10:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61e80ce149d3f3c05af95966e8795c04e7c10b9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 22:16:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e9d781148d74657bad4f235c95861f2\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e2d489acb2fd464abf81f4c79b0b0ac3 : SUCCESS in 5m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/46e7f74988f2405bacb0fc6151090325 : SUCCESS in 3m 05s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/2528f191135a4cc4984535e8354d44a0 : FAILURE in 2m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c17d5cf381273f6795b07789bc7a47b31292d134","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-13 10:16:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8201eee757df3d734f5f11fe02468efa8e880fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 10:39:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26a1e5e77ea140108dbb4f2a7398a600\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/eb2dc68aa4b7485e86980d4061475b42 : SUCCESS in 21m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ac2528baf3014a1b8209d6c302928bd8 : SUCCESS in 4m 31s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/ff29944e5ffa415d9c65b1abca07b57e : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":2},{"id":"763fac94a1f33ed19306122a90b5a5f33c9795c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 10:39:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90c521c16fe51a416df0bedc6eeae068052a5e2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 10:59:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8803915dbcda4e88843e588eaaf03eb3\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b979c658d2f944e1966f1515ed9f054a : SUCCESS in 15m 30s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/45d9fc3a9e844d2e8d8038cab2e7417b : SUCCESS in 4m 42s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/8dc49132e9f34268b2350041cb1f7fc5 : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2a40f89407cb7ec8b4d861bec38797c7738a2b54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 10:59:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7e0ea48d64d54b472d365f19ca75dafaf202518f","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:03:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2200a8adbbe943e581ade5b5a00c533d\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/d25204ddc668453e9358668027784b2d : SUCCESS in 3m 17s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/5a48ad9a34264848b6f766be6506a512 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e1993ebb01827c34018b6030b771bb5613638f77","revisions":{"ad0071f3bd4ad9bb22cf4036ac61b87c07f2c052":{"kind":"REWORK","_number":1,"created":"2025-11-12 21:49:24.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/81/966981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/966981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/966981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/966981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/966981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/966981/1"}}},"commit":{"parents":[{"commit":"1458bd29b5acd096ea1c3b4201a389e6d9e8170f","subject":"Update StarlingX docs promote job for R11 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1458bd29b5acd096ea1c3b4201a389e6d9e8170f"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-11-12 21:49:01.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-11-12 21:49:01.000000000","tz":-480},"subject":"Grafana: add core graphs to rackspace-flex","message":"Grafana: add core graphs to rackspace-flex\n\nWe have core limits in these providers, add them to the graphs.\n\nChange-Id: Ice255368db795897c3f620edbad35b57a3e7f2c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad0071f3bd4ad9bb22cf4036ac61b87c07f2c052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad0071f3bd4ad9bb22cf4036ac61b87c07f2c052"}]},"branch":"refs/heads/master"},"e1993ebb01827c34018b6030b771bb5613638f77":{"kind":"REWORK","_number":2,"created":"2025-11-12 21:50:35.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/81/966981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/966981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/966981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/966981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/966981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/966981/2"}}},"commit":{"parents":[{"commit":"1458bd29b5acd096ea1c3b4201a389e6d9e8170f","subject":"Update StarlingX docs promote job for R11 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1458bd29b5acd096ea1c3b4201a389e6d9e8170f"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-11-12 21:49:01.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2025-11-12 21:50:30.000000000","tz":-480},"subject":"Grafana: add core graphs to rackspace-flex","message":"Grafana: add core graphs to rackspace-flex\n\nWe have core limits in these providers, add them to the graphs.\n\nChange-Id: Ice255368db795897c3f620edbad35b57a3e7f2c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1993ebb01827c34018b6030b771bb5613638f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1993ebb01827c34018b6030b771bb5613638f77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
