)]}'
{"id":"openstack%2Fproject-config~970193","triplet_id":"openstack%2Fproject-config~master~I2749b2635e1358801aefa0c3836f8dd8f8d42369","project":"openstack/project-config","branch":"master","topic":"monasca-retirement","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-12-09 15:55:32.000000000","reason":"Change was submitted"}},"hashtags":["monasca-retirement","retire-monasca"],"change_id":"I2749b2635e1358801aefa0c3836f8dd8f8d42369","subject":"Cleanup Monasca infra","status":"MERGED","created":"2025-12-08 22:30:35.000000000","updated":"2025-12-09 16:09:26.000000000","submitted":"2025-12-09 15:55:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970193-monasca-retirement","meta_rev_id":"e0f7f0cccc779ec4704a2632a936b63c06562c8d","_number":970193,"virtual_id_number":970193,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-09 15:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-12-09 15:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-08 22:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-12-09 15:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-08 22:34:42.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-12-08 22:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-09 15:46:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"0263a033cb571a2dd50129dfb24f2fc092abdb51","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-12-08 22:30:35.000000000","message":"Hashtag added: retire-monasca","accounts_in_message":[],"_revision_number":1},{"id":"428f6c74cd7ffa6ca36da30271140a1e0d118ee7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-12-08 22:30:50.000000000","message":"Hashtag added: monasca-retirement","accounts_in_message":[],"_revision_number":1},{"id":"b03a10a8135d547bec2eb066435ebfdfc3e799b4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-12-08 22:34:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a27135974b30e2cef33d842cb0e509dd6bf9b77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 22:39:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f89f6f2fddf43c594fc7a63d5be8007\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9f02d6a644114b0d9e09bc1b4b2f284c : SUCCESS in 4m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5b758813eb074bbd9f82d020420b59d7 : SUCCESS in 2m 22s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a38ebc0791e746169b5b32c7eb8fedee : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c4afe99973f021d99fde5812c92f38905454d539","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-12-09 15:46:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7256a70bced92774d51baedf2cb834a7de8b524","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 15:46:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c694d8cac3390cb0585516f0afe4f9cffb08180","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 15:55:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71ca750332bb4ecc876ee51f290de5f3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5c67b4ce7823411d8f87ca79a6045fa1 : SUCCESS in 5m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8c47fed7c153438f97cba61a0dce9754 : SUCCESS in 3m 44s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4e0cec3cb60747b28f0d6883a4024cb3 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c96593ebd14cb2b5e659a916ade6987e3899b934","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 15:55:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e0f7f0cccc779ec4704a2632a936b63c06562c8d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 16:09:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3016e74bf2da47eb89690286276efd20\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/ae48ece609364e2e8bc3f9e4a0d5df9d : SUCCESS in 12m 45s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/bfba75f9f228436bb59a93a5f447a8ba : SUCCESS in 4m 58s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/89384e234bc24ba9a4d79cee5c293988 : SUCCESS in 5m 44s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/65c9f4fe6a324d388cbf6f117b8bf93c : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50454525945c5ec7a41fce36f4e466cae7a37fa9","revisions":{"50454525945c5ec7a41fce36f4e466cae7a37fa9":{"kind":"REWORK","_number":1,"created":"2025-12-08 22:30:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/93/970193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/970193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/970193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/970193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/970193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/970193/1"}}},"commit":{"parents":[{"commit":"9d4f68b9aa6cdee1a0c8b04c5601a42559063be3","subject":"Merge \"Add SeaweedFS App to StarlingX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d4f68b9aa6cdee1a0c8b04c5601a42559063be3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-12-08 22:30:17.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-12-08 22:30:17.000000000","tz":-480},"subject":"Cleanup Monasca infra","message":"Cleanup Monasca infra\n\nThe project team and all its deliverables\nwere retired from OpenStack.\n\nChange-Id: I2749b2635e1358801aefa0c3836f8dd8f8d42369\nDepends-On: https://review.opendev.org/c/openstack/governance/+/953671\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50454525945c5ec7a41fce36f4e466cae7a37fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50454525945c5ec7a41fce36f4e466cae7a37fa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
