)]}'
{"id":"openstack%2Fproject-config~972508","triplet_id":"openstack%2Fproject-config~master~I343134381ce03076497f03d639ea967050af8927","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-01-12 15:21:46.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-01-12 16:47:50.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-01-13 18:00:28.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-01-13 18:00:28.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-01-13 18:00:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I343134381ce03076497f03d639ea967050af8927","subject":"kolla: Introduce a version of Ironic core two tier structure","status":"MERGED","created":"2026-01-07 15:58:21.000000000","updated":"2026-01-13 18:07:14.000000000","submitted":"2026-01-13 18:00:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972508","meta_rev_id":"85b35909fc0be3d5f2a2c27be579ec342d278b18","_number":972508,"virtual_id_number":972508,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-13 18:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-01-12 16:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2026-01-13 17:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2026-01-13 17:49:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 15:59:54.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-01-07 16:08:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 16:14:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-01-07 17:00:58.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2026-01-12 15:08:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-01-13 17:49:59.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"6630ad435f44a6928e13f66dfa9486c4d3cd34c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-07 15:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e8e51b04bef8202efda788b0d54d2752505cb74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 16:08:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f8b09a738994c9e9dd76e7ec6b775e2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e005eee1a42741bc9f5a1f49e8242464 : SUCCESS in 3m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f4c28a79c893459ea2c091d0afd3f469 : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6fade48e896218d92da36cbf85379158088fb65a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-07 16:10:18.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6d0688176668d08655d35a9e603b355c3c1beb5d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-07 16:14:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a33bd1ba25255ae84cd2ffdd633182c203906249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 16:26:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eab0d4f1306a47748a8be754eef27f91\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e207b0ac59a8413e92097b0bf50ba145 : SUCCESS in 5m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/aaba7499c4444d67aac9b42ab96b994c : FAILURE in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ddd2cad65706d10ba972be3b48bddb287d57e6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-07 16:27:52.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9305d08ccab57b3973ca344b5980aedda90905a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 16:38:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e60c552d6304309a86204c07536cc15\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de57c8ab67334431a4e5062c91525b5c : SUCCESS in 5m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5413591c199442eda24797683c81d23e : FAILURE in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"6c2be15485222f0421c6c55057ca38b0a8d49a36","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-01-07 17:00:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"945d9d945a79feece0e6c9e69b061302c2e5d255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-01-07 17:31:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9126e140d437a4619894add4c8792b4598c6d781","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-07 17:32:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd48e31f86e3b5b136fcc5b65cd531b0c8df8f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 17:39:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e4d80dea6b49a2af8208c7dc484b43\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d608f3180a454981b4ae4a78c7ec586e : SUCCESS in 2m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f70cc21f91354b91a53116f0a0b51798 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"b926177e7afe825070767175747bb9eeb9a76c7a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-08 08:25:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9afe59a8b3379be6f9ea73242cced3721c2bc4da","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-12 15:08:34.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"940a19598c4f2986de337c99e2d55dae07fbad08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-12 15:16:52.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1e0ff79dcfe54364c64759eb8a0209367f1c92b6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-12 15:21:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4f29a5bd28549341a42284e826df48c7f748a11d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:25:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41d391f200dd4b37ab68ff46fa5ad131\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3fae0db8f4f44e69b23b4ea37ea00fd7 : SUCCESS in 4m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/de73365ce2044f24ace53e690b4add93 : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":5},{"id":"a96f4f4b37c484b3b44b8805b8767d12c3d465fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-12 15:34:45.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"00ad02435f6dbdcae59f353e5595754d1153aece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:45:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bcd0c1d97d64d60aa2240e0bca7932a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a00dca22cb2c4f5884860d9ddf69c6e4 : SUCCESS in 4m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f84b5c252176423fa592449c0f116520 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":6},{"id":"6452ab10c56d8f24046031388fa6d9bb8923bcc7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-12 16:47:50.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1c535413e05fdff3426ae5f218c17e2c4b640661","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-01-13 17:49:59.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7cf6f6b14cf63f26bcfe7d3a9af597d1bfdd1aad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:50:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"874089f8b3fe937bc1cf081223b5c9a65624937c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 18:00:28.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba30d1458a744f9b16fa760b66fa9fb\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/329fa7c084b24feaabc9d3a85e3a5153 : SUCCESS in 4m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ee5933d1f37940edbce6f23cbaab5312 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"c5309b38c6a4536c05cc4bb209f012b68f0694dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 18:00:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"85b35909fc0be3d5f2a2c27be579ec342d278b18","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 18:07:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9638238228c34eeda3ee31ca9023aa67\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/095de1c2dbab46d293defdb78cfc0a8d : SUCCESS in 6m 30s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/1043cce3aa0f4aa99c81ec705f969ec4 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"72a6fa35dd26da4a87fd34ab6a13f9d23ce7a8e6","revisions":{"faa42a6abb9effbfedd5ee3f8f0c4bc805b4047d":{"kind":"REWORK","_number":1,"created":"2026-01-07 15:58:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/972508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/972508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/972508/1"}}},"commit":{"parents":[{"commit":"b2aa57a8ab87aa12122f7dd5b955266a944d15e3","subject":"[neutron] update dashboard with new jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2aa57a8ab87aa12122f7dd5b955266a944d15e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 15:53:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 15:58:11.000000000","tz":60},"subject":"kolla: Introduce a version of Ironic core two tier structure","message":"kolla: Introduce a version of Ironic core two tier structure\n\nkolla-reviewers - +2/-2 vote rights without workflow rights\nkolla-core - +2/-2 vote rights with workflow rights\n\nkolla-core group stays as is as the most-trusted core reviewers\ngroup, with addition of kolla-reviewers for people stepping up\ninto maintainer/reviewer roles except the workflow rights to\nland code\n\nChange-Id: I343134381ce03076497f03d639ea967050af8927\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/faa42a6abb9effbfedd5ee3f8f0c4bc805b4047d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/faa42a6abb9effbfedd5ee3f8f0c4bc805b4047d"}]},"branch":"refs/heads/master"},"bbabe409acfbb8e8dfc4603d518b7c92e4bd8ba4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-07 16:10:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/972508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/972508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/972508/2"}}},"commit":{"parents":[{"commit":"b2aa57a8ab87aa12122f7dd5b955266a944d15e3","subject":"[neutron] update dashboard with new jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2aa57a8ab87aa12122f7dd5b955266a944d15e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 15:53:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 16:10:18.000000000","tz":0},"subject":"kolla: Introduce a version of Ironic core two tier structure","message":"kolla: Introduce a version of Ironic core two tier structure\n\nkolla-reviewers - +2/-2 vote rights without workflow rights\nkolla-core - +2/-2 vote rights with workflow rights\n\nkolla-core group stays as is as the most-trusted core reviewers\ngroup, with addition of kolla-reviewers for people stepping up\ninto maintainer/reviewer roles except the workflow rights to\nland code\n\nUsing the opportunity to adapt Backport candidate votes\nto only +1/-1 - neither users nor core reviewers do not\nunderstand the difference between +1 and +2 - let\u0027s simplify it.\n\nChange-Id: I343134381ce03076497f03d639ea967050af8927\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbabe409acfbb8e8dfc4603d518b7c92e4bd8ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbabe409acfbb8e8dfc4603d518b7c92e4bd8ba4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9e3a42acaecc502bba12f92bf22d671a6090059a":{"kind":"REWORK","_number":3,"created":"2026-01-07 16:27:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/972508/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/972508/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/972508/3"}}},"commit":{"parents":[{"commit":"b2aa57a8ab87aa12122f7dd5b955266a944d15e3","subject":"[neutron] update dashboard with new jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2aa57a8ab87aa12122f7dd5b955266a944d15e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 15:53:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 16:27:52.000000000","tz":0},"subject":"kolla: Introduce a version of Ironic core two tier structure","message":"kolla: Introduce a version of Ironic core two tier structure\n\nkolla-reviewers - +2/-2 vote rights without workflow rights\nkolla-core - +2/-2 vote rights with workflow rights\n\nkolla-core group stays as is as the most-trusted core reviewers\ngroup, with addition of kolla-reviewers for people stepping up\ninto maintainer/reviewer roles except the workflow rights to\nland code\n\nUsing the opportunity to adapt Backport candidate votes\nto only +1/-1 - neither users nor core reviewers do not\nunderstand the difference between +1 and +2 - let\u0027s simplify it.\n\nChange-Id: I343134381ce03076497f03d639ea967050af8927\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e3a42acaecc502bba12f92bf22d671a6090059a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e3a42acaecc502bba12f92bf22d671a6090059a"}]},"branch":"refs/heads/master"},"aee7eba78b65a0c3437781755a2511207d091975":{"kind":"REWORK","_number":4,"created":"2026-01-07 17:31:35.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/08/972508/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/972508/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/972508/4"}}},"commit":{"parents":[{"commit":"b2aa57a8ab87aa12122f7dd5b955266a944d15e3","subject":"[neutron] update dashboard with new jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2aa57a8ab87aa12122f7dd5b955266a944d15e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 15:53:18.000000000","tz":60},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-01-07 17:31:27.000000000","tz":-480},"subject":"kolla: Introduce a version of Ironic core two tier structure","message":"kolla: Introduce a version of Ironic core two tier structure\n\nkolla-reviewers - +2/-2 vote rights without workflow rights\nkolla-core - +2/-2 vote rights with workflow rights\n\nkolla-core group stays as is as the most-trusted core reviewers\ngroup, with addition of kolla-reviewers for people stepping up\ninto maintainer/reviewer roles except the workflow rights to\nland code\n\nUsing the opportunity to adapt Backport candidate votes\nto only +1/-1 - neither users nor core reviewers do not\nunderstand the difference between +1 and +2 - let\u0027s simplify it.\n\nChange-Id: I343134381ce03076497f03d639ea967050af8927\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aee7eba78b65a0c3437781755a2511207d091975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aee7eba78b65a0c3437781755a2511207d091975"}]},"branch":"refs/heads/master"},"57d76ae11ce8d8a9cb15e07dcc7e7192cbc6009b":{"kind":"REWORK","_number":5,"created":"2026-01-12 15:16:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/972508/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/972508/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/972508/5"}}},"commit":{"parents":[{"commit":"b2aa57a8ab87aa12122f7dd5b955266a944d15e3","subject":"[neutron] update dashboard with new jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2aa57a8ab87aa12122f7dd5b955266a944d15e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 15:53:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-12 15:16:52.000000000","tz":0},"subject":"kolla: Introduce a version of Ironic core two tier structure","message":"kolla: Introduce a version of Ironic core two tier structure\n\nkolla-reviewers - +2/-2 vote rights without workflow rights\nkolla-core - +2/-2 vote rights with workflow rights\n\nkolla-core group stays as is as the most-trusted core reviewers\ngroup, with addition of kolla-reviewers for people stepping up\ninto maintainer/reviewer roles except the workflow rights to\nland code\n\nUsing the opportunity to adapt Backport candidate votes\nto only +1/-1 - neither users nor core reviewers do not\nunderstand the difference between +1 and +2 - let\u0027s simplify it.\n\nChange-Id: I343134381ce03076497f03d639ea967050af8927\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57d76ae11ce8d8a9cb15e07dcc7e7192cbc6009b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57d76ae11ce8d8a9cb15e07dcc7e7192cbc6009b"}]},"branch":"refs/heads/master"},"72a6fa35dd26da4a87fd34ab6a13f9d23ce7a8e6":{"kind":"REWORK","_number":6,"created":"2026-01-12 15:34:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/972508/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/08/972508/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/08/972508/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/08/972508/6"}}},"commit":{"parents":[{"commit":"b2aa57a8ab87aa12122f7dd5b955266a944d15e3","subject":"[neutron] update dashboard with new jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2aa57a8ab87aa12122f7dd5b955266a944d15e3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-07 15:53:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-01-12 15:34:45.000000000","tz":0},"subject":"kolla: Introduce a version of Ironic core two tier structure","message":"kolla: Introduce a version of Ironic core two tier structure\n\nkolla-reviewers - +2/-2 vote rights without workflow rights\nkolla-core - +2/-2 vote rights with workflow rights\n\nkolla-core group stays as is as the most-trusted core reviewers\ngroup, with addition of kolla-reviewers for people stepping up\ninto maintainer/reviewer roles except the workflow rights to\nland code\n\nUsing the opportunity to adapt Backport candidate votes\nto only +1/-1 - neither users nor core reviewers do not\nunderstand the difference between +1 and +2 - let\u0027s simplify it.\n\nChange-Id: I343134381ce03076497f03d639ea967050af8927\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72a6fa35dd26da4a87fd34ab6a13f9d23ce7a8e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72a6fa35dd26da4a87fd34ab6a13f9d23ce7a8e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
