)]}'
{"id":"openstack%2Fproject-config~973060","triplet_id":"openstack%2Fproject-config~master~Ia0154bcdb2a64559b26910c51d3a76904df6685c","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2026-01-12 11:08:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0154bcdb2a64559b26910c51d3a76904df6685c","subject":"Update openstack-mirroring secrets","status":"MERGED","created":"2026-01-12 10:40:51.000000000","updated":"2026-01-12 11:12:01.000000000","submitted":"2026-01-12 11:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973060","meta_rev_id":"677e4fc3bd6cfab1550d6f1204b807bc87283229","_number":973060,"virtual_id_number":973060,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 11:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-01-12 11:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-01-12 11:04:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 10:51:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 11:04:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f7367746b33aef200059f7be340f4e6fd3bb5179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-01-12 10:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e7321147421986f346b8df39b8c6e524814e68d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-01-12 10:43:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"11217b3cab2895e9269ffaea27772a1e75fdcb08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 10:51:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/197e2450f053420091b2a0ac673fc672\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/14bca25fd378403ea4e8c4b46806cac9 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"b7fb9e284e26d2a7b6a16ba4b6441e97c16db2c9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-12 11:04:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c777184e4bd64f26f1db788a45cc3ff255050ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:05:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14da5612a9f64ab95df43d9acdabe407a75d4937","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:08:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc768160e0f54b8f95c3dc2b15199806\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/56164bb165614a21b2a4237ff9020e48 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"eff32770399d1fc2e185a76e103a9326fcc86041","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:08:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"677e4fc3bd6cfab1550d6f1204b807bc87283229","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:12:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54c18fb8ade4b01bce8dd30e2d32570\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/107c7b2db8cd4b11b341a4d18f00c028 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"64bf3518a732f48774adeba06308f65234cfbae4","revisions":{"213f47be16a1b8980f6ca200c81b466d8bb04976":{"kind":"REWORK","_number":1,"created":"2026-01-12 10:40:51.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/60/973060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/973060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/973060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/973060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/973060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/973060/1"}}},"commit":{"parents":[{"commit":"7424ef54bd42a9d6ad3237fa3df6ff370dc714f7","subject":"Update remaining AFS keytabs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7424ef54bd42a9d6ad3237fa3df6ff370dc714f7"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-01-12 10:36:28.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-01-12 10:40:49.000000000","tz":60},"subject":"Update openstack-mirroring secrets","message":"Update openstack-mirroring secrets\n\nSSH key and API token used for openstack-mirroring jobs on GitHub have\nbeen rotated, this patch updates the corresponding secrets in Zuul.\n\nChange-Id: Ia0154bcdb2a64559b26910c51d3a76904df6685c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/213f47be16a1b8980f6ca200c81b466d8bb04976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/213f47be16a1b8980f6ca200c81b466d8bb04976"}]},"branch":"refs/heads/master"},"64bf3518a732f48774adeba06308f65234cfbae4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-12 10:43:18.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/60/973060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/973060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/973060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/973060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/973060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/973060/2"}}},"commit":{"parents":[{"commit":"7424ef54bd42a9d6ad3237fa3df6ff370dc714f7","subject":"Update remaining AFS keytabs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7424ef54bd42a9d6ad3237fa3df6ff370dc714f7"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-01-12 10:36:28.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-01-12 10:42:51.000000000","tz":60},"subject":"Update openstack-mirroring secrets","message":"Update openstack-mirroring secrets\n\nSSH key and API token used for openstack-mirroring jobs on GitHub have\nbeen rotated, this patch updates the corresponding secrets in Zuul.\n\nChange-Id: Ia0154bcdb2a64559b26910c51d3a76904df6685c\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64bf3518a732f48774adeba06308f65234cfbae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/64bf3518a732f48774adeba06308f65234cfbae4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
