)]}'
{"id":"openstack%2Fproject-config~973963","triplet_id":"openstack%2Fproject-config~master~I072baf2c2a1e6585e94c5a704eed138536c3f20f","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2026-01-20 19:52:41.000000000","reason":"Change was submitted"}},"hashtags":["opendev-matrix-switch"],"change_id":"I072baf2c2a1e6585e94c5a704eed138536c3f20f","subject":"Remove #opendev from gerritbot","status":"MERGED","created":"2026-01-20 16:48:38.000000000","updated":"2026-01-20 19:57:05.000000000","submitted":"2026-01-20 19:52:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973963","meta_rev_id":"6e239beb0a0a4ff2de1d8ac1b1ea7f452fb3e28a","_number":973963,"virtual_id_number":973963,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 19:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-01-20 16:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-20 16:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-20 19:42:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 16:49:40.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-01-20 16:51:38.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-01-20 16:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ae24ec7e99888387874f9fe271f4a486b427b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-01-20 16:48:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7321f9bdd9d3a944a2b50d7560b2ded2ccd8d813","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-01-20 16:49:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5193dc4e70719c47786005cf53463f1b8668107","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-01-20 16:51:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11881d4d931b84f3c8a6f9cb6edff5a4ce6be404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:58:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6a5d300b82e4b25b091f973df84f3d3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9e35605d159b43f68fe624005f5439ac : SUCCESS in 4m 54s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/06dc5b587c704386a1c8897a3e4493ee : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a181d233da26aeb9a15d9e6be9ebe23c65a2189d","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-01-20 17:22:24.000000000","message":"Hashtag added: opendev-matrix-switch","accounts_in_message":[],"_revision_number":1},{"id":"179bf52ea9330ed5f026500955a1c5d720e5b86d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-01-20 19:42:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ed70f7d524ca0d2e81ecc63e37e079a965625d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 19:43:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3521a556c9dd5323721589a3789699673341c8f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 19:52:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/903db90f67ca48989ceeca63bfe7ddad\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/346ad2f7e47e4296806e0eac58ba4fc4 : SUCCESS in 5m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/63ded4e282084f52b26789a5fec37c25 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2ab55ea78a1c0b14ef82689858ee9b2a82bab356","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 19:52:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e239beb0a0a4ff2de1d8ac1b1ea7f452fb3e28a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 19:57:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd0d564934c45bb8c9ba8dc13a36815\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/277ed777ade54ceaae1a175f07372429 : SUCCESS in 4m 05s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/30c56334287340309625b84ed632f5ad : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50e4d89bd70fc1fd63ed245ac1279933323a069a","revisions":{"50e4d89bd70fc1fd63ed245ac1279933323a069a":{"kind":"REWORK","_number":1,"created":"2026-01-20 16:48:38.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/63/973963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/973963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/973963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/973963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/973963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/973963/1"}}},"commit":{"parents":[{"commit":"c7ee098414ba369ef30c1fcf68f2d8a7d43abe73","subject":"Merge \"Revive os_watcher role for OSA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7ee098414ba369ef30c1fcf68f2d8a7d43abe73"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2026-01-20 16:43:49.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2026-01-20 16:43:49.000000000","tz":-480},"subject":"Remove #opendev from gerritbot","message":"Remove #opendev from gerritbot\n\nThe #opendev channel is moving to Matrix:\nhttps://lists.opendev.org/archives/list/service-announce@lists.opendev.org/thread/MTJBNQUGOSYT4D5I2Z5ZCKN32QHCD52U/\n\nRemove gerritbot announcements from the IRC channel in order to\nclose it down.\n\nChange-Id: I072baf2c2a1e6585e94c5a704eed138536c3f20f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50e4d89bd70fc1fd63ed245ac1279933323a069a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50e4d89bd70fc1fd63ed245ac1279933323a069a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
