)]}'
{"id":"openstack%2Fproject-config~976496","triplet_id":"openstack%2Fproject-config~master~I1801e139af9120a790741acd5654abc652afea74","project":"openstack/project-config","branch":"master","topic":"remove-pkg_resources","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-02-11 16:44:44.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-11 18:37:08.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-02-11 18:37:08.000000000","reason":"Change was submitted"}},"hashtags":["remove-pkg_resources"],"change_id":"I1801e139af9120a790741acd5654abc652afea74","subject":"Onboard xstatic-angular-schema-form repo (1/2)","status":"MERGED","created":"2026-02-11 16:08:28.000000000","updated":"2026-02-11 21:43:24.000000000","submitted":"2026-02-11 18:37:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976496-remove-pkg_resources","meta_rev_id":"4368b7e4d27beb236b0703ee82c4566fd24198a5","_number":976496,"virtual_id_number":976496,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-11 18:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-02-11 16:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-02-11 17:34:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-11 18:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-11 18:29:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 16:09:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 16:18:45.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-02-11 16:21:46.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2026-02-11 17:34:00.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-02-11 18:29:39.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"a8c9cbf1ff72716a47c93fa2a94f3710d4d9214f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 16:08:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75d7829475ba8517d2d7e9d0b6058a94315fd4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 16:09:55.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/xstatic-angular-schema-form\" was not found.\n  All projects referenced within a Zuul configuration must first be\n  added to the main configuration file by the Zuul administrator.\n\nThe problem appears in the \"openstack/xstatic-angular-schema-form\" project stanza:\n\n  project:\n      name: openstack/xstatic-angular-schema-form\n      templates:\n        - noop-jobs\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4483","accounts_in_message":[],"_revision_number":1},{"id":"b2a60fe345a68d5e9d76606c1fbc9d4b9beb664b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 16:14:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5ff313c3044edfa3011126dbf2256e5a443717e9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-02-11 16:18:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e8f65c6e9ee31a37a1de9edb12bbc4f3626b69cf","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-02-11 16:21:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2afc7b9ced0a21c2b1481f9cb60d9666eca4a7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 16:24:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/577b094032a8483c83544b2b4f8eaea5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ee10db6add3d4baab73b4fba09d4a326 : SUCCESS in 4m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c29e9997d4ab48ceb8999ee5f83a1524 : FAILURE in 3m 06s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/b96dc32aec3d4c02a069ed86d7c3ef72 : FAILURE in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"65fdaaf765caf0daf64427fe4bde034326d12e91","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 16:35:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0d1bef65fb8d181dd5ee2a11837720d3701e7a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 16:36:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0e1739f418c1355b3a48c1fcee8e422484a3a2cf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-02-11 16:44:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d19c42a98917899f5bdf3619abd272de5ddb0cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 16:50:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ccee7ebd0224eada348dd94e0a57ed8\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a42606639de945aab37ec8e7de03a845 : SUCCESS in 4m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7a4c3857b32245db9703bb882314dd2e : SUCCESS in 3m 45s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/70acdc4cc05d400fb30bf4d96dc22f53 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"a9dcdb6b2a7ad6d58d895ee7cce9b2582ba123a9","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-02-11 17:34:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2a784b79bbad4970a3ff621b054ad970bbf83a19","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-02-11 18:29:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"409b98ec80f8b58552e25b6e1ddce5a44a1f3667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 18:29:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"16f7c729717870ea79f04df24e8a3a5845270cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 18:37:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c829eb44cae46d799f93f254fd3a53e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/db4978a6db50477498e9ecfe11aa863b : SUCCESS in 4m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a87842a24a24434ba095ed437008c626 : SUCCESS in 3m 28s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/f2a4d6250da247628c36aeeb1a19f098 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8842308effccb223176ef8eaa7a02ad1d17ca582","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 18:37:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f5993bbc07d4adb5271c6269feb22639bed2ddc3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 18:49:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efac0940b49245d5ae4da6beb2d2952a\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/76c5286d9f3e4615af80f358a2e61571 : SUCCESS in 12m 16s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/9c0ea12861964dd9866c577068df3599 : SUCCESS in 5m 01s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/3f17fda8633e493684e6ed33a7db72c1 : SUCCESS in 5m 17s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/67bbf2425edd496f847e232291d41097 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3},{"id":"4368b7e4d27beb236b0703ee82c4566fd24198a5","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-11 21:43:24.000000000","message":"Hashtag added: remove-pkg_resources","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d889d2cfaf6fcb095c579eb29f5fb9b848b8a7a4","revisions":{"31d98c8bab735445ef4901f1cf5e6dcfdd250013":{"kind":"REWORK","_number":1,"created":"2026-02-11 16:08:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/976496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/976496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/976496/1"}}},"commit":{"parents":[{"commit":"4c49c30dc7c3d01ee9bfd0d0831a35d20789ab9a","subject":"Merge \"Add back volvocars repos in volvocars tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c49c30dc7c3d01ee9bfd0d0831a35d20789ab9a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 16:05:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 16:08:06.000000000","tz":0},"subject":"Onboard xstatic-angular-schema-form repo","message":"Onboard xstatic-angular-schema-form repo\n\nopenstackci is already a maintainer on pypi [1].\n\n[1] https://pypi.org/project/XStatic-Angular-Schema-Form/\n\nChange-Id: I1801e139af9120a790741acd5654abc652afea74\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/31d98c8bab735445ef4901f1cf5e6dcfdd250013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/31d98c8bab735445ef4901f1cf5e6dcfdd250013"}]},"branch":"refs/heads/master"},"ee5967657fec7465a220f5c2d99c2f5dccc72b64":{"kind":"REWORK","_number":2,"created":"2026-02-11 16:14:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/976496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/976496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/976496/2"}}},"commit":{"parents":[{"commit":"4c49c30dc7c3d01ee9bfd0d0831a35d20789ab9a","subject":"Merge \"Add back volvocars repos in volvocars tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c49c30dc7c3d01ee9bfd0d0831a35d20789ab9a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 16:05:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 16:13:21.000000000","tz":0},"subject":"Onboard xstatic-angular-schema-form repo (1/2)","message":"Onboard xstatic-angular-schema-form repo (1/2)\n\nopenstackci is already a maintainer on pypi [1].\n\n[1] https://pypi.org/project/XStatic-Angular-Schema-Form/\n\nChange-Id: I1801e139af9120a790741acd5654abc652afea74\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee5967657fec7465a220f5c2d99c2f5dccc72b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee5967657fec7465a220f5c2d99c2f5dccc72b64"}]},"branch":"refs/heads/master"},"d889d2cfaf6fcb095c579eb29f5fb9b848b8a7a4":{"kind":"REWORK","_number":3,"created":"2026-02-11 16:36:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/976496/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/976496/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/976496/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/976496/3"}}},"commit":{"parents":[{"commit":"4c49c30dc7c3d01ee9bfd0d0831a35d20789ab9a","subject":"Merge \"Add back volvocars repos in volvocars tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c49c30dc7c3d01ee9bfd0d0831a35d20789ab9a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 16:05:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 16:36:15.000000000","tz":0},"subject":"Onboard xstatic-angular-schema-form repo (1/2)","message":"Onboard xstatic-angular-schema-form repo (1/2)\n\nopenstackci is already a maintainer on pypi [1].\n\n[1] https://pypi.org/project/XStatic-Angular-Schema-Form/\n\nChange-Id: I1801e139af9120a790741acd5654abc652afea74\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d889d2cfaf6fcb095c579eb29f5fb9b848b8a7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d889d2cfaf6fcb095c579eb29f5fb9b848b8a7a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
