)]}'
{"id":"openstack%2Fproject-config~976612","triplet_id":"openstack%2Fproject-config~master~I4b785ba20f7ee29379b2ac77f8dbf08ed0644ea2","project":"openstack/project-config","branch":"master","topic":"remove-pkg_resources","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-02-12 15:39:29.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-12 17:44:32.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-12 17:44:32.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-02-12 17:44:32.000000000","reason":"Change was submitted"}},"hashtags":["remove-pkg_resources"],"change_id":"I4b785ba20f7ee29379b2ac77f8dbf08ed0644ea2","subject":"Onboard more xstatic repos (1/2)","status":"MERGED","created":"2026-02-12 12:29:39.000000000","updated":"2026-02-12 21:21:55.000000000","submitted":"2026-02-12 17:44:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976612-remove-pkg_resources","meta_rev_id":"516ba5b3af2d8c05e0b1f947eb112a431c6ad5ac","_number":976612,"virtual_id_number":976612,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 17:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2026-02-12 15:39:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2026-02-12 16:02:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-12 17:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-12 17:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 12:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 12:37:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-02-12 13:04:39.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2026-02-12 15:14:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-02-12 16:02:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-02-12 17:37:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d1db17549013853109cfbaeb60753e8a9b0a163c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 12:29:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc43120ed815872a2a4c38d6d15103074f3db421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 12:35:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99bb29e5f3a14d88bf0f6f4a6ee639a4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bef347ad450041d09996d7eed1f29ed8 : SUCCESS in 3m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/951fd2e54c70492381120d997ac7a65d : FAILURE in 5m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ddae6f39954b466dbdec21f3e0932eb3 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"65434de6eac2bf16646674b057aeaa056f482085","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-12 12:37:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3718ca9d104a4b5b016b735a7d49f750f426a383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 13:01:12.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"06baeab2b4e8829fc8ee622a841aa2fe5ae3c6f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 13:02:48.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"78b30d78b5d2aa95498c0dd7a57ef51d9515f129","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 13:03:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c4058a7724b742ae2c96b6d1ee9a38f125324cb5","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-02-12 13:04:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"786c371bfbfe03d8b7129a33bbe82e3c7b3f1bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 13:11:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e381c812216f436d82218f78c50267c1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/109405bfaf1a4cc4b178619ae2fa290c : SUCCESS in 2m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0ead6d5a423642cd9b447cc9d88a7f72 : SUCCESS in 3m 13s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c9a7210ea4e3419c89d331f9d948ca31 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"57df5dd0f32d6e337f58b4aec018a51e2384bad7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-12 13:30:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f8d99ee2d7174db56f321907852e6a7df3bb273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-12 13:54:49.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6352ee6e04519492d6364056c73ace37a92ebded","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-12 13:54:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"db70152896cb3f7b85bc1e0e1695d32499c0f269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 14:06:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f16d3b55545b4191a103fba51340efbd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c4ae32048d84416f94ea71ef4af8dcba : SUCCESS in 5m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/31fb70e22a944c439b35425f36d7b533 : SUCCESS in 3m 54s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/63a76afa18714b779e0e8795e1efbfb2 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":4},{"id":"006b6900d18608c1561423b6dbb7c4776337d532","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 14:30:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"53a8c4caee90cee6b449018c9c5de464656a73e4","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-02-12 15:39:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"de7804cc90e31d0f85318e6e94285f8ed94386d9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-02-12 16:02:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"61a21619d6d175b782719a0b98661684344f7c81","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-02-12 17:37:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"04a72b9e05d7ac480fb84b93e5a6c9ab114fed4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:37:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e5fb2877379d47a64b03cb8d0a8c0fbe75ae600f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:44:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c9a30d4f8454a618c1adf3c4bf0edfa\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fcd93153db2f4d1abeabca1065ab42dc : SUCCESS in 3m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9d0f4d347e0a4038a544884b8a18de57 : SUCCESS in 2m 19s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/25600f58f48340a09bfd6a73d9c28e46 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":4},{"id":"cd2f0c366009f0e4269b9192a4479e62a4115af6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:44:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a6209868ae44ec1909879beab8755ee4e820f89b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:57:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d146bc560e841a48ccc666e74807a12\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/65ae906c66184c7191dc8c3d8215786d : SUCCESS in 12m 35s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/ed1874a311214812929f742a3ab5af0e : SUCCESS in 4m 56s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/26fecdac24d147e9bf66be0140bbbbdc : SUCCESS in 5m 19s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/d70917dd4c914cba84916ad74005febf : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":4},{"id":"516ba5b3af2d8c05e0b1f947eb112a431c6ad5ac","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-12 21:21:55.000000000","message":"Hashtag added: remove-pkg_resources","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3189ccd1b7689806c61734d2398da4fc98724ee6","revisions":{"6c1112f5bd93fa1f7886815c8b948cdff9411c9d":{"kind":"REWORK","_number":1,"created":"2026-02-12 12:29:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/976612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/976612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/976612/1"}}},"commit":{"parents":[{"commit":"a84e9bfbed88464742e2d54ef9fd348690c55667","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a84e9bfbed88464742e2d54ef9fd348690c55667"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-12 12:12:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-12 12:29:07.000000000","tz":0},"subject":"Onboard more xstatic repos (1/2)","message":"Onboard more xstatic repos (1/2)\n\nOnboard xstatic-objectstore and xstatic-tv4. As with\nI1801e139af9120a790741acd5654abc652afea74, openstackci is already a\nmaintainer of these projects so they just needed  to be reconstructed.\n\nChange-Id: I4b785ba20f7ee29379b2ac77f8dbf08ed0644ea2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c1112f5bd93fa1f7886815c8b948cdff9411c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c1112f5bd93fa1f7886815c8b948cdff9411c9d"}]},"branch":"refs/heads/master"},"90bcf42f13cce468a4b3957fd51c82dfc1a55f91":{"kind":"REWORK","_number":2,"created":"2026-02-12 13:01:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/976612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/976612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/976612/2"}}},"commit":{"parents":[{"commit":"a84e9bfbed88464742e2d54ef9fd348690c55667","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a84e9bfbed88464742e2d54ef9fd348690c55667"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-12 12:12:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-12 13:01:10.000000000","tz":0},"subject":"Onboard more xstatic repos (1/2)","message":"Onboard more xstatic repos (1/2)\n\nOnboard xstatic-objectstore and xstatic-tv4. As with\nI1801e139af9120a790741acd5654abc652afea74, openstackci is already a\nmaintainer of these projects so they just needed  to be reconstructed.\n\nChange-Id: I4b785ba20f7ee29379b2ac77f8dbf08ed0644ea2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90bcf42f13cce468a4b3957fd51c82dfc1a55f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90bcf42f13cce468a4b3957fd51c82dfc1a55f91"}]},"branch":"refs/heads/master"},"767930e84c22faa9ca837af5b6a70e1614409926":{"kind":"NO_CHANGE","_number":3,"created":"2026-02-12 13:02:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/976612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/976612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/976612/3"}}},"commit":{"parents":[{"commit":"a84e9bfbed88464742e2d54ef9fd348690c55667","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a84e9bfbed88464742e2d54ef9fd348690c55667"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-12 12:12:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-12 12:12:53.000000000","tz":0},"subject":"Onboard more xstatic repos (1/2)","message":"Onboard more xstatic repos (1/2)\n\nOnboard xstatic-objectstore and xstatic-tv4. As with\nI1801e139af9120a790741acd5654abc652afea74, openstackci is already a\nmaintainer of these projects so they just needed  to be reconstructed.\n\nChange-Id: I4b785ba20f7ee29379b2ac77f8dbf08ed0644ea2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/767930e84c22faa9ca837af5b6a70e1614409926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/767930e84c22faa9ca837af5b6a70e1614409926"}]},"branch":"refs/heads/master"},"3189ccd1b7689806c61734d2398da4fc98724ee6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-12 13:54:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/976612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/976612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/976612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/976612/4"}}},"commit":{"parents":[{"commit":"a84e9bfbed88464742e2d54ef9fd348690c55667","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a84e9bfbed88464742e2d54ef9fd348690c55667"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-12 12:12:53.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-12 13:54:49.000000000","tz":0},"subject":"Onboard more xstatic repos (1/2)","message":"Onboard more xstatic repos (1/2)\n\nOnboard xstatic-objectpath and xstatic-tv4. As with\nI1801e139af9120a790741acd5654abc652afea74, openstackci is already a\nmaintainer of these projects so they just needed  to be reconstructed.\n\nChange-Id: I4b785ba20f7ee29379b2ac77f8dbf08ed0644ea2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3189ccd1b7689806c61734d2398da4fc98724ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3189ccd1b7689806c61734d2398da4fc98724ee6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
