)]}'
{"id":"openstack%2Fproject-config~978334","triplet_id":"openstack%2Fproject-config~master~Ic984404f2e09f36e781f9651b831b577bb8af494","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ic984404f2e09f36e781f9651b831b577bb8af494","subject":"Remove googlesource projects","status":"ABANDONED","created":"2026-02-28 18:02:48.000000000","updated":"2026-03-04 22:09:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d1fead1ef36b756f71e34615d9c6bc3e755746e6","_number":978334,"virtual_id_number":978334,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-28 18:07:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-28 18:07:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c179db47b7bb99e00420cdfeca98bc5909a697b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-02-28 18:02:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae70b06612ddfd8098520c49e9cea38eb1f0d0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 18:07:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe25f50db264427a22e6e0a46ca4076\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4431a60d30634f26a1dca4368b2ac467 : SUCCESS in 2m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4585ec4ad2f4412382d609751aa8d643 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d1fead1ef36b756f71e34615d9c6bc3e755746e6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-03-04 22:09:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ffbc8c2b0055a0f996f8c38a1c8479e72204baa","revisions":{"9ffbc8c2b0055a0f996f8c38a1c8479e72204baa":{"kind":"REWORK","_number":1,"created":"2026-02-28 18:02:48.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/34/978334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/978334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/978334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/978334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/978334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/978334/1"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2026-02-28 17:57:57.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2026-02-28 17:57:57.000000000","tz":-480},"subject":"Remove googlesource projects","message":"Remove googlesource projects\n\nThe Gerrit server at gerrit-review.googlesource.com appears to be\nthrottling access to the /info/refs endpoint which Zuul uses to\ndetermine what branches exist and their contents.  Zuul waits 30\nseconds for a response, but the remote server usually times out.\n\nThe result is that tenant configurations with these projects take\na very long time to complete while we await the timeouts.  The\nreconfigurations complete without the information about branches,\nwhich means the next time we start without a cache, or it is\ninvalidated, we won\u0027t have any branch information for these projects.\n\nChange-Id: Ic984404f2e09f36e781f9651b831b577bb8af494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ffbc8c2b0055a0f996f8c38a1c8479e72204baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ffbc8c2b0055a0f996f8c38a1c8479e72204baa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
